)]}'
{"id":"openstack%2Frequirements~976248","triplet_id":"openstack%2Frequirements~stable%2F2025.2~I035715bc63fa7b929617007f4cd04bc7a7d3aad3","project":"openstack/requirements","branch":"stable/2025.2","topic":"do-not-merge","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-02-10 13:39:19.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I035715bc63fa7b929617007f4cd04bc7a7d3aad3","subject":"DNM: CI health test","status":"NEW","created":"2026-02-10 13:11:36.000000000","updated":"2026-03-30 10:16:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2792b209002cc553b50d52149cdc77d127d742ad","_number":976248,"virtual_id_number":976248,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-30 10:16:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 13:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 17:58:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-03-25 11:58:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"6694e3b0a2f05c085d00271439bbb356c16aa318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-10 13:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"430f84f0287236ad579dcfc847a6b550eec20d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:39:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb6ecb96688f4e71878469dc29038f3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92579aeecff43149d80b459ec58ad2b : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c2d59d82e29488faa34ccf2fddb0c78 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f5bab9b3fc34722a84dcbb9db09d323 : FAILURE in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8a12f835d924c8e8ec1eeca7676cb60 : FAILURE in 3m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60b9e5c2b04d40299985b8fa8af90be6 : FAILURE in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfadb382f6604f87a8ab88bcefe12795 : SUCCESS in 3m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7976c1d7f3b941d298fa9ff0765214e8 : SUCCESS in 14m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f221c3632ec4661808e8cd491ae6b7b : FAILURE in 3m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ce28b549cd3b4ec9bcccf9d1b14faab6 : FAILURE in 4m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/118ead7757774d0783b985b182d6932e : FAILURE in 8m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c8fb5b73fecf40699ab76df5da9ebf44 : SUCCESS in 4m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c21a2914850f4451a47d3f3b5727495d : SUCCESS in 5m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/09ddd46d33784ddda9260d162514b385 : SUCCESS in 4m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9173e17e24db4a3685d10a7197a4e330 : FAILURE in 5m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5b2922a08fb14e999bd62a29d1419b32 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1fa260c1886649a1b37b6c791d5a55ae : SUCCESS in 5m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/22ed7f7381774d41a648b9acf3017c06 : SUCCESS in 5m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4dbe39aea29743edb1ff7a49d8a44d9d : FAILURE in 1m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1f7f8db40d04ff584b860fe7a837cf6 : FAILURE in 4m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8af2adea17094ad38c96804fc08688a7 : SUCCESS in 10m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4011a4c4eb29459faa210a42f031f33b : SUCCESS in 12m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/29ceb9abcae546b692b627281115b9d6 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/471cbf4abe9e4660b5003999c23c5ecc : SUCCESS in 10m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4f9ffd4b3dd84c61b53163445022fa00 : SUCCESS in 4m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6a324abfc8574f2f9d19a70071995864 : FAILURE in 3m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ae6cd00b504d405386515080bbe08acc : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8e816943699244d9b87a0e76d2269a77 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/641720bdc2da4becba7b1212c41ee6f0 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/45342c8526c84be8bbdef51483b3b218 : SUCCESS in 11m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7f5379efe1b84e4584f9757b3ddbeac0 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/913cad130aa946a48944ccd2361fe58d : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6767ee8a3dcc4b148bfcbbc11520f637 : SUCCESS in 24m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4b422159f30a4fa5b6c082abb0f6507c : SUCCESS in 7m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0c1a94e3e07b4fcfbf83e69c3f1a845d : FAILURE in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/795b0db84ce54b069465d3cd21ffe0bf : SUCCESS in 3m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b83005566be24c02b53213f2b9a31785 : SUCCESS in 20m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e729e3b9020a49558540c977d1368234 : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5de7a58e85d45c7be748b27338f96d9 : POST_FAILURE in 13m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dfd7ac4162447b0be72d373af5debaa : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7224ab50e34ad3c04d690e9baced82efa190412a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-10 13:42:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e0592f0773ad42a6cd0e26e7be98dcb2a6a1a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 14:20:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e355f3dfa1274bf4868c2e66cb8879c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5ee8cab2794e6db6f98c21d250ee4c : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86e25f82aa6445789c00b6b86eccdc54 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8eed5f4f0c5d4bc4b01cb4bfdb7d3c0a : FAILURE in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd32d2c4ebed40b58ea3862d0bd8d4d1 : FAILURE in 6m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88c84ad205d54aaa8f17fc6f485bedaf : FAILURE in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4286edb251742719b8cf2dcbadabac8 : SUCCESS in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b8022257a5f54e27ba57a1ab688f9cbc : SUCCESS in 12m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8aae86ab10254033aa4b1ab6c436c0b8 : FAILURE in 7m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f0cd427099f74e32826e339eaeb89f60 : FAILURE in 5m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/98ff1f5adf2b4c03a4e336d8112badfd : FAILURE in 5m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ecb47d5c54644a30acaddb5041786df6 : SUCCESS in 4m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3e465a2e0444428aa9996be206f50a23 : SUCCESS in 6m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/602a11cc396445a98d6f0b66b53642f7 : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/72df8392949541a082b966ef66381228 : FAILURE in 7m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/889b5048afa646e98f74694b0628c31f : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1c9989cf87bd4736aec34f73ca46f039 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e4d3f89b18474c9cab7852fedc442a20 : SUCCESS in 9m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/17f1486090d242b09f880e963e396d11 : FAILURE in 1m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47b20803068944eaafbaecf235fb98ea : FAILURE in 5m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d59bb1e112934ab291ec949e9afe18cc : SUCCESS in 9m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fe334be1548a4efd85fb38dbddfb5a82 : SUCCESS in 13m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9b01494e764f48549fb832023fd19fab : SUCCESS in 3m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/127be48648e542e7ae16f8fa308dce88 : SUCCESS in 11m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2e6eb8322918437083f6edc21b5ed2a0 : SUCCESS in 2m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a6928e2cd5b54fa5863192bf809db26b : FAILURE in 2m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/04c51908be6e4934995aa68818425326 : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4c45799dae2d480da026b95f6c33b02a : SUCCESS in 13m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b746e447d09a430b85c1273eb2b6dbc0 : SUCCESS in 7m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/53f95b70c4ea42e79ae2bbf409df9bf5 : SUCCESS in 7m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3d0de9c4a0814300b9114482ad6c611c : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b863f973a8a4a0284de43d9fa3141c3 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ab1e07fad9b4101a5f149aedee79b63 : SUCCESS in 31m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bb4d368857024c7ca3d84ba4f5a491a6 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b192196cd0cd47b99c3d64b2eb5a6e3d : FAILURE in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f600e20e90b0424a813c9aaeabc293d6 : SUCCESS in 5m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e0525e61724a4024ad804d43111a75be : SUCCESS in 14m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8b25e1d7db16400fb920f559e89b2b44 : SUCCESS in 6m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f8096e01e7d4a4aaeb33fbebdecad4e : POST_FAILURE in 7m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3b8535ffe184409bfe27edef1a7fedd : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1e0c34c75ce5b3a97c0e26e9228b398688aa64f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-10 14:20:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c959b25e80a75e47599ef733df6f9def39659df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 14:52:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83d9db7c9f8c4dcf809bfc86961fde51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076acfbfe9c84d4e8deaeeaa3e2980bd : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a281eb88375646b6987a0d8327becc7e : SUCCESS in 2m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26bc2831545f4a8ab6adf2cff8d577f3 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d590c7af06248c8aa44ef1294c62e4d : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/124174b6f0664e4f924b7fa1415a8a49 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53629a39bc7e4d19b59de8beabdffb98 : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/67ba0a64c0714cbcabeee60b8919e96d : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/078237b4b10b4a10bf5db2980680f867 : FAILURE in 4m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8edec03ec5fc4b0c8f54c6d39469424e : FAILURE in 4m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b7d1006c111b492c8040de92c1d7a7e5 : FAILURE in 5m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b49737dd18064a93bd66b8304336bcae : SUCCESS in 4m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8205135608c940cb92a6bfb8b7fa3915 : SUCCESS in 4m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/adf72733f36e47538c410e8dfb5889d2 : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6f4078570026433bb4fd0d60bf789258 : FAILURE in 4m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8045107aeacf4403bfe921889d7d0a08 : SUCCESS in 5m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1eda9f4e1ea94791adbfb7038ab7ab31 : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/178079176b004c41b4cd2ddb823af1a9 : SUCCESS in 7m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/379ae8ddb2204c8ab77bfe12418f3e26 : FAILURE in 1m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d65bef66e48447dfbad7580e2a82915e : FAILURE in 3m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4101b6c4e9a3481c9bcc715dbd122e7e : SUCCESS in 7m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/07f99eaadd2b4602b12686d86811613a : SUCCESS in 9m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e72017a88a1a400fa8ee3dd1db990e87 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6cee621bc46f44a1b37e5bd1c8cf1228 : SUCCESS in 9m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2f30cb26fbef4e00a73b956831e63edc : SUCCESS in 4m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dfbb4802fcb04a9eb897b3d0c14b8d06 : FAILURE in 4m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a7265a9760e44cfc9fcaa7cee0b868f0 : SUCCESS in 10m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/520eac1c6eca4f4faf217cf28223d006 : SUCCESS in 15m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e4a74ffd34184e2d949a2bb6f0f2e3f2 : SUCCESS in 4m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3aee82ee088e4e2e988efd79a39a0783 : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7f4eca028cad48c394319382acb6be2d : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/070c28b09adb437f8af4b2e019c3098e : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/256f8aec3bea4eb0ae573bd2bdcfcbbc : SUCCESS in 26m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d3116b6d5f4f46d2b9af06e8d7677cb1 : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b9e87da3309f49a9ba50abcefe4582e0 : FAILURE in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14842a7849bb4f25a3b259fda6ab0f44 : SUCCESS in 3m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f96a3fccccf24d70b86eef4f7a903698 : SUCCESS in 17m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/868761527e7041d79aa5aaaf0dce961d : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ff6b3b4540b490fa659c1a8b1755c31 : POST_FAILURE in 14m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa694c124ced4584b2eb582f7ec46f3d : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fe7b760bc93923e0c9a8a598e5774c57f127d808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-11 20:54:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":4},{"id":"f1598cfa33e2363c8689f3c4730bc3d34d3bfc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 22:54:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baaab068b91b496da11ee14a3acd6379\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a85019d57a4eb79af30402955ebf94 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1c8898e422341c3852bf9651a512f0e : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68d6872edf5341e89900b3c3db9030e8 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d48196305fc4b1492b595b6dc450234 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8160a1642f0642d6963823e6613a6798 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68c6ec5016fb4de39bc014b61c3c3558 : SUCCESS in 4m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dc6f3e3873bc4ae6ab8835d6f9b7725d : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38832da907724d6c9f5e847e61e824bd : FAILURE in 7m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3a23af8ec35a411c9d6540713f4212ac : FAILURE in 7m 00s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/07ae7590b5d64f3197f5d25e57080495 : FAILURE in 6m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f77ab20be0534251bbbae030bee153a2 : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e13dd00238204443802b91a440e7865d : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/33ee6bdd18eb4605bda2b483c58aad3a : SUCCESS in 4m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6bea6133363e45d9ab0e74301e8043c7 : FAILURE in 5m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b6906eaf233042e99cfd45192dc79c5a : SUCCESS in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f54daf93998d49b5ac906d406b7314c2 : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/500152b7a42344f2a6226552a5ec138e : SUCCESS in 3m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/44342ff42b2d4481b3bec4a19dcd0a0b : FAILURE in 2m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26f004f1457d42e3943cc2510ad7cb24 : FAILURE in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/66bb40f0feb14440bf0d1f1f286f0bbb : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e4bb213d04c149ed8ec8f1032585c9de : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3f1afe6d23fb416d8efa05fb642197e0 : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/18037fd81096423daca5d9885dc10937 : SUCCESS in 9m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4821dad458284208ba3c148cac7ff913 : SUCCESS in 4m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0de6ed1fcaac4f56b394823ca65eca38 : FAILURE in 4m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d1c672b6cacb47f8bfd1c3637d2fd2d1 : SUCCESS in 10m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cfed44db9e224cb4a6f3d391b0ce17e5 : SUCCESS in 22m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9831c0f118ae40e5a4ec2a0e35f01c33 : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b0d0716776f54afbbec079103afd3e26 : SUCCESS in 12m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a61ecb1b7285471d8f82933882456ed0 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7109bae0c46f4156a19391fe77624cdb : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25d0614cccd3462e80b12821c6e170b7 : SUCCESS in 26m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/289335ff2e2b4edbade7d70f623ec890 : SUCCESS in 3m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d5a47fadc33b492cb294514871a407c2 : FAILURE in 2m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f01eec2e40544ca4a633fa19f9c8a6be : SUCCESS in 5m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6663cb45076647178b15a54ab7a36971 : SUCCESS in 7m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6af549940e8248e0a523fb4a328ecc1f : SUCCESS in 5m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2b547121d5641b185ace5302779719d : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dfd538893b24230b22c3f4ac8bfe2d4 : SUCCESS in 1h 54m 26s","accounts_in_message":[],"_revision_number":4},{"id":"73f8761c47cebce8f69544181dd988999839c25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-16 16:41:44.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ac8ae34d60deda6703c257077a5343df657a53cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:36:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05b45324a2054dbfb8adba3bd8a78a43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b63c22bbe3a4a0385481af7e627847c : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/567594a057c9461f859489b69e02f1c4 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8c7bff8a9694d8dbd0e26ae71b115d3 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bee3579d09847f78f2bfa8ab9af784a : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e9e42e57a7c497290142e8d8b80eedb : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9af50e959a4d40e9bd607208ce23d7d0 : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bf180a66e32149c9a2aab18d8dd1f740 : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ee694c7c970948ffa1a90f4a858aafd2 : FAILURE in 4m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/89d2d6742d114e538ecb2771e07d7e00 : FAILURE in 4m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a11f869c694c4489aad32775cf81330b : FAILURE in 6m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3aa34eba4a6b44889e745ec3035c33cc : SUCCESS in 5m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1431fdda1452452aa58d5513ee2a8faf : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e2be2b855b504862899cbaf767f00dc8 : SUCCESS in 4m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cdae88a6c19e4b2abca562eba7e719ca : FAILURE in 5m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/52df5c52a0854f0e94c18cc679d9fdf5 : SUCCESS in 4m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a71d8d4eac5435ba24a53678a358bf1 : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2db3b074e08a48c4bba4468ed286eb51 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ae7914571574453980a12e393751d71 : FAILURE in 2m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/971bcfd4a5a24c4d9693c9e7ce919094 : FAILURE in 6m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0dc7ca6d4ae74398858d60847aeeb3de : SUCCESS in 10m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5d1d2d6a7aa9475695d5eaa48d0f0532 : SUCCESS in 7m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b52887b10f4c46b5816f00d8ba6d0970 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf5bbad83dd84209a16637c71d09cd2e : SUCCESS in 11m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0b0260668d374c508d6df8f9f8c9ae4b : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a253fb4fdee1446b8bead2885975e340 : FAILURE in 5m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0ae7dbd6376a41018df968cd87336ff4 : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5decc65ac5164d8b898278f4bd81548d : SUCCESS in 19m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2fa28fc816504061bae426890442f38c : SUCCESS in 4m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c2345234c3ff426cbc00229c23d20f9c : SUCCESS in 13m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5d649637c0424566aefffc067dcd7012 : SUCCESS in 3m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/180a982abc72446682180b0f5918a40a : SUCCESS in 2m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b8231e2c8ed46b4b62999f561d0d15c : SUCCESS in 24m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a505800960094092bd01ac7eca2b74ef : SUCCESS in 7m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9f580baa5d5f43728c658bf395bc736b : FAILURE in 2m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4825fd378b14103ae155ab9cb8dafcf : SUCCESS in 3m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4070bb17618e4681a0a1083a7b3899f3 : SUCCESS in 19m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2d3b66eab547472795b7d04703657ea4 : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365ae0ee87824af0b627aacdf5c87435 : SUCCESS in 59m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/638a11353ad04ec9a9ed6b01ce5c880d : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5d42ce8b06e66bf7e9b7787bd4216c5e687d602f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-24 17:58:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"322d149baec6500b86c59a19b291b4c1d65dda32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:24:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13be19b5ef3a43a99bd765b477102298\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fdc80bd9e64ad1b8d6a6a6130e7be0 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/876fdc4b84a24feca5ee57a5adcb0b8b : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8a893ac3e784f5f837bc1bb79bd185a : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89024a2dae894d5dacb34da1473aac28 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30f286625f2e4b569435e17a6f492c63 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01b114553ae7493d93788a20cd7f5037 : SUCCESS in 2m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8e89dc0635cd47458aba1da9c5f04307 : SUCCESS in 8m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9e94aaccd064481ea35c19a1904369de : FAILURE in 4m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bcc1e241d902438790da2f0d09d7437f : FAILURE in 3m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/82895c2d2eb34de8b408b257795a2258 : FAILURE in 10m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a2d0f3180580474f9e492fd960cf38b0 : SUCCESS in 4m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fa223d8e51cd4af8af267394817164c2 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2f1ce33b2c9d4b39b1b2cc4ac5aa0d20 : SUCCESS in 3m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/87cce59da7514dfeaadc911e5af30a55 : FAILURE in 5m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e0d0b50f9d4048a399e122f6f81a5b26 : SUCCESS in 7m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a2b66c528cbe48e48c93365537a49158 : SUCCESS in 5m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/896e89e2dd824f40875417b1856080a8 : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dc3fadd92ce9402696474815ff2051b6 : FAILURE in 2m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59116fb4119e45b083bcfa65d30caf0f : FAILURE in 5m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d3b84cb19a2b4a8789c8bfa4940add6e : SUCCESS in 8m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1369e78265dd4a3093883cf799a1e428 : SUCCESS in 13m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eb68fee5500c4fabbcc202821da47953 : SUCCESS in 2m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e7b4070806ef47fb84fb57167dac8b2e : SUCCESS in 11m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d778594a44e6482082090dd1d586585a : SUCCESS in 4m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1fad96a5f349421a89a558bfc1a5130d : FAILURE in 5m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e6a1fa46d11041e397231546e6e109c8 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1057edc04ecf4e2984c96e0d6f444ba4 : SUCCESS in 19m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d8b4ff3604c541c9865f45f22fbe85ba : SUCCESS in 5m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/df7c42062dd848d396d2830f38395202 : SUCCESS in 9m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f089317634374d628a7d0e2464d88b8f : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1dae4533217e4d919e5cc266d8887574 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40b424e278724deeb1a30a1b9dbd0294 : SUCCESS in 28m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01e26583d185413596a10495a21a4400 : SUCCESS in 6m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dede2240616642cda9ebd887a5aae2d0 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90510586c672463ebac3e26a9df56492 : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b5ac7df60b9148bfb1083bff76ef2734 : SUCCESS in 17m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/968a6729b9bc44429ef3199d12ae0e35 : SUCCESS in 8m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30e29175cc9049299b03390a18bcc001 : SUCCESS in 1h 06m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2874063e215640d596e2b130af4e1ef4 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":5},{"id":"530a3186746562c352d0ab9d5962ced155cb82a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-23 15:11:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cab889c0c1613442832efe68750de002fdc6f713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-23 15:12:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"59c16b88c895c5e93d9127291f8279c6c3186fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 17:07:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95a6750a90f84269b6f7f4a2c0c999b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d09634f01894d62afada94d60861e7c : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8efc77092c8a427d9e803ad7a539ee3b : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4e63396a2a34a3e9284e8ce9ddf13cd : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6928a86f92784956b0822c4da1181e28 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d26e67ae15c54587a46c1dc60cb016d6 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2967a8c59f7a432688008dfd199be236 : SUCCESS in 3m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c856c65fd9c464f82265c4862cfe1a0 : SUCCESS in 4m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bd2a9eee82e3436f9347057bf32a675e : FAILURE in 2m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9bbfff4694d84325804e69a27026b315 : FAILURE in 6m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3d37027613114275b9b5236f3a382213 : FAILURE in 6m 23s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f7ff2686cd054597b5bf5748020be632 : SUCCESS in 6m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cdb2baf9bdef48bc8db7c94da485dc69 : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/480543b408404e71abdcebc762a624e1 : SUCCESS in 6m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b3516a78a44144dea9039264a09e290c : SUCCESS in 11m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e7b7af687db24856ab7f9b2a14b37eda : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/78e1ccf594304f1b8544c9113cc25a88 : SUCCESS in 6m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5447d64f06ac44089b1268d6c39e74fe : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4bd8e31245e144b48692d465a488a60a : SUCCESS in 17m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46e1e881c1ea4f9ab2e162c4117bc794 : SUCCESS in 10m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c98f28af95d540e8bf3c4d971ee4d5d8 : SUCCESS in 10m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc97bbcea5f841589daa311cf3af7caa : SUCCESS in 13m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/58be7b8fa82c4385b445722af2b39666 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/391a5b7416ad480e95cfb61fdf38c7f6 : SUCCESS in 12m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/76588d600afa4ca8a8791fbfdb281877 : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1145228262844ee6bac44ccfb2b831a2 : FAILURE in 2m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/22725865819b4f0a93645c8352bd07c1 : SUCCESS in 11m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/da8738d835e948ea94416db3de1ba71a : SUCCESS in 18m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6fa9f88ffd7b47cf888ce9b90e3e0ccf : SUCCESS in 7m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/572baf58ac1d44709dfd444be6cc2680 : SUCCESS in 9m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/464616fd0a0645b1a2c98e5932c898e6 : SUCCESS in 3m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/140b30424c2d4d308c5bd255d866cd92 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d6c18a2cb374986a6b7cfe3edc3fc26 : SUCCESS in 27m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7217efbfc8ee4cafb20afad2f443c7e7 : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ce041caf2d3547a4b257fb198df01002 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e851263b854a4ff082c93e313467493a : SUCCESS in 7m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/12610344652b4f6190899870bfa033da : SUCCESS in 21m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/37cf1d39371f4ab88cc99ab100dec254 : SUCCESS in 5m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcf50001653d4f9cbaaec951227c64ef : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7977f200a39e494fa4a5762d007ab9c5 : SUCCESS in 1h 49m 27s","accounts_in_message":[],"_revision_number":7},{"id":"b4002d93c04252508687302e2b2888825d4506e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-24 12:16:03.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"32f7068f975c96c3a50e0a18be3d0d56533d388d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 14:13:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31f868c9c4f84700bf943f469bbdfd8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d553d7a67143bda8a543f615676d56 : POST_FAILURE in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cee0aa8529e8429fa67ae0c503ef6ce9 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85ef1716aa16491ab40d8407b726f109 : POST_FAILURE in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f45c6299c13c45bba514924ce7d4588a : POST_FAILURE in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a00c416c6714873a4475436e9c41f1c : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/943efcbda0494ef69ca6b2ecf93fcdac : SUCCESS in 4m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f97e3a35fe8c45c7a557e7d60011d02f : SUCCESS in 10m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5a1db9aaf0e74610873bd90ecab57c4f : POST_FAILURE in 4m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ad9db9ac04344814965502f85443b52d : FAILURE in 5m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0ff51f2ecd294039ae2836c6430bbc98 : POST_FAILURE in 9m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2d3c83a8b3d64de394859fc425131357 : SUCCESS in 7m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a275c9b1fc3f49fc867cb843d77e1d9c : POST_FAILURE in 6m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a1c8987351c24497a7f9f578b3e0cea2 : POST_FAILURE in 6m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/44b808f6f72a4089a8a63b514833f50f : SUCCESS in 11m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e577a6b3f70649ca928a8394d4ecb3d1 : POST_FAILURE in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9a9ba9628c6d4fe4932003034c04b98a : SUCCESS in 5m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5acacbdc35f64e57a65a4e353faca142 : POST_FAILURE in 8m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c6b2377327f94dce941db9a286cb1b61 : POST_FAILURE in 16m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/647c22aad88c4c18bafebf571586d1e9 : SUCCESS in 4m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a6e15c5d9b047d6951f26a03881e142 : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aba93f9022e14143b8daee3579cd5c9e : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9299ec1cd9ce44c291385ddce3e3023e : POST_FAILURE in 3m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/92dfdad3088c4283ba9d6f2c420cae2b : POST_FAILURE in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/15ccc9be917645a6846547247d751859 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/533cbc829b1f4619941f0b72ecbf4472 : POST_FAILURE in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/be4cfec7eff24b039f9aa10516cf83ca : SUCCESS in 9m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3057133bf8a44a88ade1830f172223dc : SUCCESS in 23m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5ae826ee0aac439b8b21b9b175805e2b : SUCCESS in 8m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f7147933d87f4784b7e62f7c29b3d52e : POST_FAILURE in 8m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9978170d35fb4e798da26292068c9f5a : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff630affa2ce4fefab992da08a832025 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aecd8a4cf667467aaf761ade51838057 : SUCCESS in 26m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2409a1476f794e1698dba14a0f198450 : POST_FAILURE in 3m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3aff3d1f80554f1aba9580591b04b16e : POST_FAILURE in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12087ec92b5c4c70835ea09045fb0c7d : POST_FAILURE in 7m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1ac5841a42ed48d7abd027b7ddb943cb : POST_FAILURE in 19m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b404bbe9f48f42f7b65e5b474d64fcdf : POST_FAILURE in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2908621ca3ae4f8ab81e5394c67f2105 : POST_FAILURE in 1h 00m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a83fc08823ac4abc84a0c6a4ab9a1a07 : SUCCESS in 1h 50m 45s","accounts_in_message":[],"_revision_number":8},{"id":"b7539eaf84ed555da75a3dbf1c1fe9f935eb3cae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 11:58:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b935b9dfe0f3bad5277e036a60b54f9d8fa6d00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:39:43.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0641d945b8fe4338be0db4d65406dad6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c6b8f4d41747f0bc3c26fd95663bf4 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/583488db8ced4e2a8b16d9f3100a031a : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c98902e45f724596aca725fda68af099 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a7c9d3581ab42a58dbd838594734029 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9631e82ca854003b517de13af91615a : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc23fca541824147b12ab632539df4b9 : SUCCESS in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6a92e35574ab4864afc1ee6f492e9eb5 : SUCCESS in 18m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e4a66165af254fd2a8970be768181b62 : FAILURE in 4m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2754e1f46f67469bb5e31f90bdc2366d : FAILURE in 5m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/febc7fc05143498a968e749d805c545d : FAILURE in 7m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/75fad49f50b042a6b50360781008156f : SUCCESS in 3m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/de7467d458f84e1c8dcda4fbd395d777 : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/70b01a6c971c49b0b9def26af8b8b2f3 : SUCCESS in 3m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b2923db9de1b40c189ede5daa631fdc0 : SUCCESS in 8m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c366de6e89e64723add63359876a9d93 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c0972a327cde4416a3bf16a0a6ed497d : SUCCESS in 7m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b7102e6ceefe42e9accbb878e734c595 : SUCCESS in 7m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7fa307705f4e4ce393cd304f1553d011 : SUCCESS in 19m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5307935478544c478056441921e475d2 : SUCCESS in 5m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/469bacfee34d4bb18c50666371565176 : SUCCESS in 9m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1d75f1d6269f456a894092c8cb54f780 : SUCCESS in 12m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1cc203ea17424abca0fa319353133f8e : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7c4575ae91bc4d868f4ee4e58f0e8e21 : SUCCESS in 11m 33s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cd22791c4aa0475db00ae782759c96bf : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/36e183b9af7349eab5cfbc8a822ccbc3 : FAILURE in 2m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d7bb11e3c8454e239d8adfc19ca940b1 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fa0969daeefc4ae6b80ea35478e5c8a8 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e952ad2f264f4f8bafb1a035d3451e16 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1561c3251cda496f9dabd9adcf846c6a : SUCCESS in 16m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9cc6a0934be4aa88f6caefb3de433bc : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47dc584753254ca1962295a30b74b471 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b21f20f98e944db88e18f93b4cf08160 : SUCCESS in 25m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5d011e34d55346639e1c43e09425017f : SUCCESS in 4m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6cd43a66b34b4df7b3dcb0c34a04ed14 : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24b9c0fdaf994763b2b3e3d95da886f7 : SUCCESS in 8m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1ff3f8e03b9d4ba6af87833dd5706db5 : SUCCESS in 13m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2fce9d72d5fc4b82a4717dddcaa96458 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb583f1ec23548e78e5e0a34d41afa64 : SUCCESS in 1h 02m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62ecc6e92c8f484a99a493c797390ae9 : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":8},{"id":"d3261f31ef9d4f5504608e6b084dccfb0770bf66","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-30 08:42:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2792b209002cc553b50d52149cdc77d127d742ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 10:16:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89bbc49d5ff241b69bfdd916c9cff949\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a130da5189d4995aadc3036083047b0 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a14e781c6a24215bf953417b7e651b0 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcabd7a60ffc48e6982960e5d91bd03f : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc39af8be13142df8cba13c4aaa20b97 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a7f039212dc4bee9cdacb3c7208b7f5 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e1b45bf62d24c3d99241187f60efabd : SUCCESS in 3m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d4be1deb8a75487a85dfc0dcbeec963e : SUCCESS in 9m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/87c57ae1c2014ae4b63ab813dcc905a2 : SUCCESS in 17m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b5f6f26d2b5141eaab29ded7b4b6b71e : SUCCESS in 18m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/df4d735688fc41c69ab8d65ff1489e01 : SUCCESS in 24m 46s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/65745216a9f3412c9b56327b567618fb : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/901388175129494aa9d8d4bd9c7d6493 : SUCCESS in 4m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/21e588eb27a54cf185c43b993ba97262 : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/01039942ef534b14b3a783481d1354fd : SUCCESS in 11m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/90a57bae0bdb43c690748139ed250bcc : SUCCESS in 4m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a01a527592344b5b44975a61de3bb9a : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/972aa81dd53a456989dfa93cb50649e5 : SUCCESS in 6m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ef20a47513b64e97b1ea4f9a7d1234a6 : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e55154fd79a446c58b9a08173b517b42 : SUCCESS in 6m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7de3b14d115a45e4ac4124f495bfdc48 : SUCCESS in 10m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/74e434bed709405387b2398df0628f85 : SUCCESS in 7m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/04f84745334346539aa83f402deca02d : SUCCESS in 3m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5b340caa38f1409293974a4d84a4275f : SUCCESS in 11m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7eaa084c15044c49ad90c2fcb925cf5c : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/df1ea15c0a454c43895ed7d5bb7fbe5e : FAILURE in 2m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/510ce517062b45bba29a697489c39783 : SUCCESS in 13m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a08ffb1298084ff3a7f9691ad1d1983f : SUCCESS in 19m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9ee28e6d59c2407b98f2a151c340cf7b : SUCCESS in 9m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9af8a5402ea64c788d0a19a87cc0fdd1 : SUCCESS in 10m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c077adcb9fb44708be2fcd865aa3027e : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5cfb990f63084886a78c02fd697ad988 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3e52cf2dd2c4cc1be32ad5d98f92818 : SUCCESS in 27m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/50f12e5375fa40ec8f3bb6e901f7526a : SUCCESS in 3m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6986bcfc213a4549abce94f09a04c073 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31791d3a00eb47d0892ff32a84b22962 : SUCCESS in 8m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/da635bdb9a6c469da5f7b13a11a21570 : SUCCESS in 18m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/59dc3e85e16e4efaa89072629a539b56 : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb0d605a136047d785569d598447ce85 : SUCCESS in 38m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/128ea71bafad4d94aed8210f8d19aec2 : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d16f2fefdf947a9d7b707eb0d5598fc0e9eafe58","revisions":{"39e6f5aaba547f8cdfef2819af0a8338802815ca":{"kind":"REWORK","_number":1,"created":"2026-02-10 13:11:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/1"}}},"commit":{"parents":[{"commit":"3d40384730d72e0be170e3bcad584e1f10eceabb","subject":"update constraint for python-ironicclient to new release 5.13.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d40384730d72e0be170e3bcad584e1f10eceabb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:33.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e6f5aaba547f8cdfef2819af0a8338802815ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e6f5aaba547f8cdfef2819af0a8338802815ca"}]},"branch":"refs/heads/stable/2025.2"},"bb3cbdcd56d78474c7a481926b751b98df9ec8d5":{"kind":"REWORK","_number":2,"created":"2026-02-10 13:42:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/2"}}},"commit":{"parents":[{"commit":"3d40384730d72e0be170e3bcad584e1f10eceabb","subject":"update constraint for python-ironicclient to new release 5.13.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d40384730d72e0be170e3bcad584e1f10eceabb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:42:43.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb3cbdcd56d78474c7a481926b751b98df9ec8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb3cbdcd56d78474c7a481926b751b98df9ec8d5"}]},"branch":"refs/heads/stable/2025.2"},"1c5329cd01a1436c63524bd08ac830249df095bf":{"kind":"REWORK","_number":3,"created":"2026-02-10 14:20:42.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/3"}}},"commit":{"parents":[{"commit":"3d40384730d72e0be170e3bcad584e1f10eceabb","subject":"update constraint for python-ironicclient to new release 5.13.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d40384730d72e0be170e3bcad584e1f10eceabb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 14:20:35.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c5329cd01a1436c63524bd08ac830249df095bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c5329cd01a1436c63524bd08ac830249df095bf"}]},"branch":"refs/heads/stable/2025.2"},"26006c2f43f9e1bd7e9ff956598758851b8e426d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-11 20:54:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/48/976248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/4"}}},"commit":{"parents":[{"commit":"ed0f971305bee6b48fb803e479c62ab0118c15be","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed0f971305bee6b48fb803e479c62ab0118c15be"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-11 20:54:16.000000000","tz":0},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26006c2f43f9e1bd7e9ff956598758851b8e426d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26006c2f43f9e1bd7e9ff956598758851b8e426d"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"},"9c3a5b515e4b7c56737fe6fa16f681f4f06fb5e6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-16 16:41:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/5"}}},"commit":{"parents":[{"commit":"ad78d8346bd5e3abd4ba5a901867baab6581f0d8","subject":"Add pyproject.toml to fix install issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad78d8346bd5e3abd4ba5a901867baab6581f0d8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-16 16:41:44.000000000","tz":0},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c3a5b515e4b7c56737fe6fa16f681f4f06fb5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c3a5b515e4b7c56737fe6fa16f681f4f06fb5e6"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"},"866a2ed9d428260e25b7ee7a7d615f18e4cb90f2":{"kind":"REWORK","_number":6,"created":"2026-03-23 15:11:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/6"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-23 15:11:08.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/866a2ed9d428260e25b7ee7a7d615f18e4cb90f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/866a2ed9d428260e25b7ee7a7d615f18e4cb90f2"}]},"branch":"refs/heads/stable/2025.2"},"65c5d2e68d6f02e12535ac2544c1d1ec837c0731":{"kind":"REWORK","_number":7,"created":"2026-03-23 15:12:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/7"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-23 15:12:42.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65c5d2e68d6f02e12535ac2544c1d1ec837c0731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65c5d2e68d6f02e12535ac2544c1d1ec837c0731"}]},"branch":"refs/heads/stable/2025.2"},"d16f2fefdf947a9d7b707eb0d5598fc0e9eafe58":{"kind":"REWORK","_number":8,"created":"2026-03-24 12:16:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/976248/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/976248/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/976248/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/976248/8"}}},"commit":{"parents":[{"commit":"29b5866a9627867ec873a9eb56aa10a5618eae3a","subject":"Merge \"Drop Monasca\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29b5866a9627867ec873a9eb56aa10a5618eae3a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 13:11:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-24 12:15:56.000000000","tz":60},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: I035715bc63fa7b929617007f4cd04bc7a7d3aad3\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d16f2fefdf947a9d7b707eb0d5598fc0e9eafe58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d16f2fefdf947a9d7b707eb0d5598fc0e9eafe58"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
