)]}'
{"id":"openstack%2Frequirements~976640","triplet_id":"openstack%2Frequirements~master~I74176be6671fd0319950e61827d175ba6f77a47b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I74176be6671fd0319950e61827d175ba6f77a47b","subject":"update constraint for os-traits to new release 3.6.0","status":"MERGED","created":"2026-02-12 15:26:38.000000000","updated":"2026-02-16 22:48:07.000000000","submitted":"2026-02-16 22:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976640-new-release","meta_rev_id":"f0043bd3431394c1eec4f91fe5549384c99a53c2","_number":976640,"virtual_id_number":976640,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 22:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-16 20:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-16 20:56:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 16:45:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 17:14:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-16 20:56:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ff77d8a2ef1e36395a2193e6ab6a7f9b08779a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-12 15:26:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2766b1ea7fdfacff1683ed2581d7164d77300252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:45:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da4fa3c817034059bf6a446ca53d1318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c5b8c87ea04740b625d498a6b064ca : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9823dfa061cc428a9fbfcd2c96bc40db : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/795b7c3641304d02b2f0aff4c6cd631c : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae740940672f435ba4bf2926f54fcf9f : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8aca77bf32454cbca64472d786e68662 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83ee20bc433e47c3acf867675cf1a468 : SUCCESS in 2m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97dbb01838464d6d9b92ce0d998c76b4 : SUCCESS in 4m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59375000778c424a814d0543b00c40b0 : FAILURE in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0e723bd705b44ad082374c2cadf6d5b7 : FAILURE in 6m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7642c7f731cf44d4bb5a1a07ae1c402a : FAILURE in 4m 03s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/55199027640e4adb946ce7f5ff8f317e : SUCCESS in 3m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6dcbfd07b2794d06b283c1fab8880381 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3fc1881ca7b04b0bad92ad07b9b72dda : SUCCESS in 3m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1f5d0697b2c243f28488ce2a14c62f5d : FAILURE in 6m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6de5f8129f3d4a77be02d9527da66142 : SUCCESS in 5m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9317d6cc50b344e28a1badccc50abc68 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a0a5de072991447882b0abce879a833a : SUCCESS in 7m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f73364a9982a43788a1ea90bb33e2bdd : FAILURE in 2m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8335dacca62d4381a43c03aed527f808 : FAILURE in 4m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/752caaed90cf47d58f2689c8c9c8fba6 : SUCCESS in 10m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/40f75d2ac1eb49518da68da02843e558 : SUCCESS in 13m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/147365fb9eff42688f9bbca6028bcae9 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e92ceac647294c72bccc2fb8b9ea7cca : SUCCESS in 6m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d941b7a834174d88a6fef510acc205a6 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a4e1f022514d45e0b06df5356f82ca14 : SUCCESS in 5m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7442fcff035f4f548dcf707edf01daad : SUCCESS in 11m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c58b59a9440649fc804f6dd840cac373 : SUCCESS in 23m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba8c48f950ae4fb3a47940492cbad6d1 : SUCCESS in 4m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2dab2d8d928444c7a2656d8ebb8f19e2 : SUCCESS in 14m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/092a783883fe4b3eac71cb1864cc59dc : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a69e559176544020839d785cb1b38df6 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4252b9f97e354de193b52ce23b2efc12 : SUCCESS in 30m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f87f0020c4ba48dcba1df2af1a49dc4e : SUCCESS in 7m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/341ecacb9f7e4ce68739f01f13ad0edd : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf48d15397e849488c7b201c811ab679 : SUCCESS in 6m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/123189ae165343a78b9ff93f03240b77 : SUCCESS in 20m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a47465dc06f94df38aeec7a28ac5fddf : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab1a5127ec3c43c7a534d7bdae459466 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59932d50c4ef4a12b85c52497be3b0f4 : SUCCESS in 1h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c616e5f8e166283659d9b89a097bb370adb1a69a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 17:14:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b145a8ff9c374e9f76a31e4077569853c048416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:04:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ac388d8d0e4934b3829c1bbd601a29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ddd4e14556b4275ba9ecbf2092cabfb : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bcc276309d421e98f25f870f0f6e22 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1701a1ea85d149848bde2d8cda6c68dd : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5648d05288754c83801345edf935aee3 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abeb4eb2afe449a8972abe049d6f6236 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6af401481c0841478786e5db1cc52170 : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/43ab5fca8a8d4ec4a4b084c91d4d3752 : SUCCESS in 6m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3b5e5c43049f4958985e23b27cdd51f8 : SUCCESS in 6m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fc583fa4e71f498dbcf3e4fe1ab880e5 : SUCCESS in 10m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/157994084d55421c9fd6c6f42adf0f31 : SUCCESS in 9m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0c4f7a28a6bd415f8782945b980dad64 : SUCCESS in 7m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1ac5cfd06fed41e9ab262a5b6dc931ad : SUCCESS in 4m 35s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/162a129854144b23a2ccd96dc4873783 : SUCCESS in 4m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c57c00bbc0dc42f68946791852afe4c7 : FAILURE in 5m 55s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ace13d6ab3be46e98218710401ae2027 : SUCCESS in 3m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90ca75a306fa4233a2b55a3206fa34da : SUCCESS in 3m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fb0064b80ec24ae7af8005aeac789146 : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/638b72b6058542339616fac8de158fc2 : SUCCESS in 15m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce4af0f4d03940239a011361f5d566e0 : SUCCESS in 4m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9500ac6596d54ebe99c46210cea2be6a : FAILURE in 9m 22s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8d6e59fa7b35459ebe59d75aaf7597bf : SUCCESS in 9m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eabd53241cdc4826ba1facead7d71e29 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/26e0493ad2154a15bb227a9524b724dd : SUCCESS in 5m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/529918485521419b826aba3239a1c27c : SUCCESS in 3m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ec9b5d88b26473bb64d58bc0cba0714 : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/61f69ce3872943a0b48e285ecc1e45ca : SUCCESS in 12m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ba47aded4b5a47e7857a749937171384 : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/427a3904ac894e94ab0b762ff77cad67 : SUCCESS in 4m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/06749210226146fabe8bf7d1eb577f04 : SUCCESS in 13m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/20e3e9a5e5f549759f78637a6a58b62e : SUCCESS in 3m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65c83369b06d441eb793f55632335627 : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b3f450df9ef47d99e15cb9ee482db19 : SUCCESS in 23m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3b3720eb2d06468fbeb02f608cceb85a : SUCCESS in 6m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/21a1757cf3f247fd9cb645281a01632e : SUCCESS in 3m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e4bdeaf5b44459db1e9fcd0818acc83 : SUCCESS in 7m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ee1f823c0d854c2e9ec4e7c8c0a5c58e : SUCCESS in 7m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6183879d39964b81bb4ee5679d796fda : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/994f8de0493c48818e0c9062bc1465a4 : SUCCESS in 1h 06m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/217eb42ecd0747b2b7d3a599178a82c3 : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1baf2a8eaaee42c48bd31c49b0416618df042e4f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 20:56:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0c46b7d15370714a4e519c5320a6e0ebe810af0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:56:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0d5057b5cc844c013a33379d0b93655db3829e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e1330c7a684c0fafa8020b96445f44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e61cfe3a02bb403b8b8159e5f632a17f : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d81e8c1de3411db9657a0f9a2d9c80 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59f7a00b94e947f0b8e684175bac6733 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/768d10214e0b47a3853cd8b9cd7a8d49 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c481c596a9044b2ab9a9831a63efebb1 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b0c9dd481d7440cb1289e183d26d408 : SUCCESS in 2m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a78a946e7004e06baa16b91da13f73a : SUCCESS in 2m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d63899532ddd453083a666fc01e3511c : SUCCESS in 9m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a3d09a89b7234115b18932d90b16023d : SUCCESS in 6m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fb2be60e7e844155a13fa04c70d7e88f : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bc005730945547c6b799cf6c7ba57f14 : SUCCESS in 6m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4c1119b7e3ff44b193d7fb54480aa8fd : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/734de56f6bcc44fcb7598e6ca6a6366a : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f57316e1e08d439697e45cf5d430270c : SUCCESS in 6m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f6e318a17b794392ba632442631d6f92 : SUCCESS in 11m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e1a0a8e1775f4a92aa59bd743fcb9a8e : SUCCESS in 1m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f3fcf0219803450a86e06877d44a9832 : SUCCESS in 7m 49s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d1aa30f901f34077baaac129953c9607 : SUCCESS in 28m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc1c516ac74f43b8b9c6ee658ddc30f7 : SUCCESS in 5m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3b8d685b8aef4a0aa62d850ea77f5438 : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/00e87a6a486448838f9d9853fa902939 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/474277c76c264d67a23f9c0ca823f3c5 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18a1f2d4bed942dfb413cc37b7830a6a : SUCCESS in 27m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1cc9d0b72cc644ffb46da78def03d5d3 : SUCCESS in 7m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe27363e10794cb2b98a3e5f3429575c : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84ba12c186e342f984c694f5686fcd63 : SUCCESS in 6m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1826d01e1af04947a1493bc3044ac7e2 : SUCCESS in 18m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/66789f2fa9df4a60835834ce64e555f7 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/407837eaa14a4ef58128d4bf230d36ee : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b27b903878147f9a6f327437a17c933 : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f0043bd3431394c1eec4f91fe5549384c99a53c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02db69ca02ec2bc8e0f565550b09ea0d8fc22c04","revisions":{"02db69ca02ec2bc8e0f565550b09ea0d8fc22c04":{"kind":"REWORK","_number":1,"created":"2026-02-12 15:26:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/976640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/976640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/976640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/976640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/976640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/976640/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-12 15:16:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-12 15:16:35.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.6.0","message":"update constraint for os-traits to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Commit: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Co-Authored-By: Claude Opus 4.6 \u003cnoreply@anthropic.com\u003e\nmeta: release:Change-Id: I3c01507b0d56f5f2057bc1c8583e38580da981c9\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I74176be6671fd0319950e61827d175ba6f77a47b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02db69ca02ec2bc8e0f565550b09ea0d8fc22c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02db69ca02ec2bc8e0f565550b09ea0d8fc22c04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
