)]}'
{"id":"openstack%2Frequirements~976772","triplet_id":"openstack%2Frequirements~master~I9b550f99df475b5eb99f4f4f6b8ba3bc57fd85f7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9b550f99df475b5eb99f4f4f6b8ba3bc57fd85f7","subject":"update constraint for oslo.service to new release 4.5.0","status":"MERGED","created":"2026-02-13 17:43:12.000000000","updated":"2026-02-16 22:48:46.000000000","submitted":"2026-02-16 22:48:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976772-new-release","meta_rev_id":"a646aa6ec5c4b2dc49a119b5a102166efe4be43a","_number":976772,"virtual_id_number":976772,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 22:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-16 20:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-16 20:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:31:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 20:24:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-02-16 17:14:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-16 20:55:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6e547841abda0262f1581cd85716391538c0037d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:43:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47141d52eff663fea398ecabeb7944462d115ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:31:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27302d72e7834fd687bb95d16d6648e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656aed0667f34901a41e79e89aae8bad : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14686413efd452187b32d2fb8d85bf2 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0db7aec1bc86447e8d43c8977edf09c1 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97c34f3244ce425cb1dd16a41f81f94b : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/842e6fa39d8a44d3a4f651903631cdcb : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b5221d0e7844375a004011d00d70485 : SUCCESS in 2m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3beb644b65544d44a1ee1b33bcccdf7f : SUCCESS in 5m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d1ebf18ed0714d1db37e4300346db7c3 : FAILURE in 2m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d8dfee00f0ad4b69b716014b720dd25c : FAILURE in 2m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2107e768b7084f49bea422e5485d0a4c : FAILURE in 3m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9e1ca25e59324278a4b4a78473655227 : SUCCESS in 6m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ad5d8b92e14d4a59a8472eec0e57e3d8 : SUCCESS in 5m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/29a9f4c1166041cfa9e93056875c73a7 : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bbfc07df3be64e96a7d8a762f52d784f : FAILURE in 2m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8cd0bbac30b1405b99b69590ddd61aea : SUCCESS in 3m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/885f015fc1b44d13b7735d0aca1365b3 : SUCCESS in 2m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3d96f282422d432289040ae758c56ed8 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/40f6ec372ead46aea14aa102b69ead84 : FAILURE in 2m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd4ee7ca4dae4bd0a06b377bfc9f7f59 : FAILURE in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/722905d8cd05447897afb24ba61a8a7c : SUCCESS in 7m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c8faa0cf1f1647b49ec147851c71e82f : SUCCESS in 14m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/50da16f8e12e4b96bdaec6923b573dea : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4fdd32a453f5469691895fe48931e713 : SUCCESS in 10m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/60a43ab21cef4c708e216eaa8d25698b : SUCCESS in 2m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7954e5824f674652ae68f5a6eb143466 : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/622b198716a241b6b3df72862c5d2734 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/60a49d8fe7f14beeb050df4707335684 : SUCCESS in 27m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4c5c1494da044e478f8c12b8256846a9 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3de7568fe74d48319c3fc831c0174a08 : SUCCESS in 13m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/461164a33cc54a5f9577283f6b205bf6 : SUCCESS in 2m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24513ab626bf40628ea14c6778fd77d6 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/651ce8bbf5354b028e78a9b488078bd5 : SUCCESS in 31m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cc80465a558e4c74863b4a268ba5efdc : SUCCESS in 6m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e412d6e32b2d42a0ae2784ed8b35c474 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af2cb960e3d94790b078366ea8f96fca : SUCCESS in 3m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/07676766a5d244f7a2f0c3c42e0b7a9a : SUCCESS in 13m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/89ca4ab9c2fa45a8b0de89866e1c4990 : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4551120f217a4b09b5717755ccede21d : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/959c5d06d8ec4c08a0a9d334688f285d : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c9011aad7f4ff2c785043dab79f2af91ca153099","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-02-13 20:24:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af9181ee4236dd5ab1477f08980cfe8697f28a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 22:14:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8c419a4b68e46b7b5a4577c85c8505f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a78855ca2fd4a4f85d3a3a5ddf6cf49 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de0a4a09e6e4c53b5980ac108c8b788 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64f0b4bde7784ee0a19ec77b6072e4ed : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e098d67e30ec47f4b21eb9a576423424 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd187ca07fca4b27a39699882aacb5b6 : SUCCESS in 1m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/172b90896fc748c0be9703677b19976e : SUCCESS in 3m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a608de8e0274fd4a1419d11bc6bda97 : SUCCESS in 7m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a041430acdff48d7a1a26175005e5fb0 : FAILURE in 6m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c3ff60c908a5428792c551cd7c9e3e54 : FAILURE in 4m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2ccb0471642e420f9d30d0201af712f5 : FAILURE in 2m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cc9bcc6dd3764aa09641e7dd9ddbcd13 : SUCCESS in 4m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/edda0d831142443bbc89913cf5ab382b : SUCCESS in 5m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bea7cd5792814252992bd9884fa20b05 : SUCCESS in 2m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2eca763d2a35444a8acb38bb2186a823 : FAILURE in 5m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a6f2ee5a789a496c8e8459d6d7f7fa69 : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/69bdfd0ffd754a0b8094e7ceffeb59bd : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6ef22f4652b941c6b1909f75dd4356cf : SUCCESS in 6m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/debc1c0a14b24dec94ec881453ae695a : FAILURE in 2m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd213770ecf04928ada770cca8820f0a : FAILURE in 2m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c0c0064d3956415eb9ec111339ad6a6b : FAILURE in 5m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d482305e06344ee0be1ce5ebe84e3c61 : SUCCESS in 11m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/33d0eba1c565484ebf3d819a9fc18124 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9f79a185a919461cab16246037b8f9bb : SUCCESS in 12m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f43b8ad194fc47b6bf47a9624d7b90a7 : SUCCESS in 4m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/eb025c2e361f4be2a7885ec4ac4558dc : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b29b2118c35b4a95826c831837251b9b : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6f18ef8377934f94ab81413ff128bbeb : SUCCESS in 26m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ddeedcf40f7041ec8f0675d6c555f8f7 : SUCCESS in 6m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/afc4f16ec75b4ad1bb754f2b00d6dbbd : SUCCESS in 14m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/114ca4b5ddbc43ba81903ccd178eaa4c : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b0687c626a64a1f9ade7bedc0b969c6 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f18dd6edd83d4a6498caed2abe6a413a : SUCCESS in 26m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/932e5e82ee8d4dc98893cdc67f6614d4 : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ed5101bb8c364b54a0fa5d72d5b78577 : SUCCESS in 2m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1773a11d031457e924a44cd527e28be : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c1a6e9995ca046ae9e144051cc46372b : SUCCESS in 12m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5efcc82f4711453bab44dfb6c7cb9e13 : SUCCESS in 6m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8fa53f3e8914f1fa0612a3103f18d4b : SUCCESS in 45m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccea20491884424aaeca333025d05ba7 : SUCCESS in 1h 48m 12s","accounts_in_message":[],"_revision_number":1},{"id":"eaf19e7dbaccd9930b53887bfbcbdcce0e3aba6a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 17:14:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a836cbc05bb91eced70a7306536545a1139fa83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:58:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb8b9b2aa8e47198adb4ede06086606\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd863758d5d49d484c687d5e40729c0 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3e25c0116c4a62937ecf7db434f6ff : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/001de57aa9df4198ab510d8d1cdceb48 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1aff2275ca344e4eb00ee36bb508046d : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95c3d49471824b75a06fa7cce1a5c3d5 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f3eb435ae7a423d97e55c017463cad8 : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0f3500b4bf64b88aa56a14bd5ef5139 : SUCCESS in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1801dae2469d413da65242c1c94ab928 : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e6f45ad8df4c417da6f8e4b8bd959b18 : SUCCESS in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6156c688703f4422b8207b556b0de60c : SUCCESS in 5m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3d59f965f2774a3fb542fe1262ff4e1c : SUCCESS in 5m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c3ccee6bd2014126a4dab39283467a96 : SUCCESS in 5m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/41381b3a3d9c4c95a92aae5e91b18dd7 : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cb96176e2bee47569451b37ca37235f3 : FAILURE in 6m 40s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/488961f725a9445c9c770e87a4f67eee : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d1e500d2ce194287b0c5bb21642f55d9 : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/911dd0c4f110475a82fb78dec6f2e916 : SUCCESS in 7m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/247e2342d82843489501fc0b63a65020 : SUCCESS in 14m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7597c3477dc74f1a96c256f2c00c1010 : SUCCESS in 8m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b8d330cec2db47b0915240f4c984c731 : FAILURE in 8m 31s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/02067d5bd6824bbf9717b5054a8d8d94 : SUCCESS in 11m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6e9831e183bf45cf927ebef274aa3fb7 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7c72d14ab05f4514b25d7a9064243341 : SUCCESS in 10m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3d39279c4a094bdab17cc86d7d2453c5 : SUCCESS in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5be3c8118cae4ead963e051a98468ec5 : SUCCESS in 6m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/323ad44cf68046f98fa6eea810a5c84b : SUCCESS in 11m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6ee84ebf328f449ab20b15b6ebb2aa3f : SUCCESS in 35m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3a48df8df4b348d3b0102ba852042cf6 : SUCCESS in 9m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f31bc0e2eada4b9fbd9a825d49f49469 : SUCCESS in 15m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/10d6b63707da4a6ebdf756c704a94918 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c65694d2e5044dd0b751ab16306cd16f : SUCCESS in 2m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/756f2c7aec5e41399201e8a84888f83f : SUCCESS in 25m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/40c87bc85a8d46769611c070905e1d04 : SUCCESS in 6m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3c43809f02654df0ac3725c223d498f2 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6efcd4c1695447cb9ea5a796b422fa99 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0bcd710380c9410bb4d5957c4471315e : SUCCESS in 21m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a0018d3d1b324c39baa780dcb70b5edf : SUCCESS in 6m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea6d03d8df7b407d829e004a3e983cb7 : SUCCESS in 51m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60f89f796d5c48738503d474a7eaa292 : SUCCESS in 1h 33m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2506f50938191a006fc2554ffab33cda5753323a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 20:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4397b1592e73c85694efbd115279937ede69700a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8c5cd0e9a60febed733495586d689ad6577f8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5b6069affc42289164d0003e77241d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e88634e3e6f4c03b9bfee481ef02b64 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc6b9cf372f48bf95b3a79a6fcc28d7 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7729752b346b4874bcd5008e0b48d33f : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c1b84caeb614f31834910d34bcf13cb : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/699dfb60ebea4a3a88085858bd3cf03e : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d791114ed54d4d018097344b040f498c : SUCCESS in 4m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/32037e016ae8427ab1722def488fac46 : SUCCESS in 4m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c6a736220c254cd49924c8790e5b12d0 : SUCCESS in 6m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fa6d4270a11d475689283a199062f817 : SUCCESS in 8m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b8c99026a8a94d8aad0e819e274d2c5f : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c7c05ac676b843788d4e26e90a28b808 : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/90aa47f20ba54635805a6d1f3b74d6c8 : SUCCESS in 9m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f86dcf903c1845679b1b71abfa306468 : SUCCESS in 15m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63a929a38e3448738a08ab15b3559b28 : SUCCESS in 6m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/70e76b65b8734f0f9bec0582b86680d6 : SUCCESS in 10m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e98155ce84b14c0d9e9a8e862ca881db : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4897a456b38940a4a5f76d42240784df : SUCCESS in 9m 52s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b2e9f8c29d1847aa9cf690313dfef5fd : SUCCESS in 28m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec5c970bfc5f472d8620264a36d44917 : SUCCESS in 5m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/95d9eee206484159925215a0757c90d2 : SUCCESS in 14m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d766a5ab381d4e7cb83b42492949d59f : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4386627e30194331a19744392c7bdbc6 : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f29c0c11db574c10b14adafe70ef9560 : SUCCESS in 25m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ca15706bf4db43c7b387740891eefba8 : SUCCESS in 4m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/14b273405e8240369a01a7dd40affc4f : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b603aba1c624c57a084acc3817e4807 : SUCCESS in 7m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a73819cf339d4abbb283dbd5d6c58694 : SUCCESS in 19m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dd8765d2e46f4c63acb30b385de51d40 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83b33cae14fc4f598f28958fa36df41a : SUCCESS in 46m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9756e7c656dd4429b866ae8076d85528 : SUCCESS in 1h 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4ee83c04200e42f75d0f556006a606c5a3755754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a646aa6ec5c4b2dc49a119b5a102166efe4be43a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d42ab882d7b409fba0a2069abdf5ac2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ed7f3069fb44899aa4b65009d926f2 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35f9656fabe4a7c98d93f17873b603cf213c9e04","revisions":{"35f9656fabe4a7c98d93f17873b603cf213c9e04":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:43:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/976772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/976772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/976772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/976772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/976772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/976772/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:33:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:33:10.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.5.0","message":"update constraint for oslo.service to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\nmeta: release:Change-Id: Idf7fbd038e226e6123f5895bff01a6c36ae3ae2e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I9b550f99df475b5eb99f4f4f6b8ba3bc57fd85f7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35f9656fabe4a7c98d93f17873b603cf213c9e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35f9656fabe4a7c98d93f17873b603cf213c9e04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
