)]}'
{"id":"openstack%2Frequirements~976774","triplet_id":"openstack%2Frequirements~master~I18795a43f74ea0e5601ffad8b911eaf63b453204","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I18795a43f74ea0e5601ffad8b911eaf63b453204","subject":"update constraint for XStatic-roboto-fontface to new release 0.8.0.1","status":"ABANDONED","created":"2026-02-13 17:49:48.000000000","updated":"2026-02-16 16:22:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5bcfa6d81576c620f9a307387629149eb4fdacd7","_number":976774,"virtual_id_number":976774,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:14:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b0a4fe3f1d424db9234a70b594e4d3a48de26a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97968b4c5c5174a7e4375c8908f869ea5c9d9341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:14:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d87089162984669bb40f65ff8371f88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee78d8a323c414c9ae62cde9dd18613 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6e3b2bf97c431eb4e6f9e08db75b56 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c8c433f2a9647d787f7523293a3ed49 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e839ccb446840cca150f0ca3418bf79 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e265fde2f5864e23ac6fa90512ad2acf : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/311da33b4c9848cc810edacb13c0685c : SUCCESS in 2m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7074e1efa09643e39fdc1d592ada9bb3 : SUCCESS in 4m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9658e84ebcd944d4accab469032deb40 : FAILURE in 3m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c506e6972ef846009f2ca4be8b07c74f : FAILURE in 5m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b75b7499bf464bb8976f03930d54c627 : FAILURE in 6m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8a872bb53b6f46f7993f1db439698c7e : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3f4c6c9b96f048eab359eec17e4727ad : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0775e5ff11a447eb80af691003cae06c : SUCCESS in 4m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/24f4a80b34ba4dbdb3263b883f653aa5 : FAILURE in 5m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3025d677382f4252a03b04b535317e55 : SUCCESS in 6m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/da6fc37b5b3b4479953ff494dbb734b0 : SUCCESS in 6m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/858a0d3eca5446ee82b873fec1a2c84b : SUCCESS in 6m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4408a92685934c4992c5cebbd8e30bac : FAILURE in 2m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3dbc31edd4b4ce1aef7674c5a07a496 : FAILURE in 3m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/65d6c12a22a446ef86a94347eb02c586 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0ea45ee227c6495aa1643be97854211d : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/09c5b8fd7dea45a88829d390b145d89f : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a349f82e85ef4d198fe9f863957ba97a : SUCCESS in 12m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0a5aa606385b4a8eb60424882e342b33 : SUCCESS in 3m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/788f356df0984264a2f8df58b12e4115 : SUCCESS in 3m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cbc6e73f7ba04f7e94cd78fa8bace35b : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/31ebc74b507544a9831433c07c1af3a3 : SUCCESS in 28m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc5560a7630b4d11949d9909a3016552 : SUCCESS in 8m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e608c2511a0467fbe637abde81ab4c6 : SUCCESS in 15m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f9a7be7ae08346d1b0a7b80c7673d01a : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c3a30e55ed34b3ba03e5983a18810ce : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3da2bf3528214123a015ac1916a0ff2e : SUCCESS in 23m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ac4bfeaa72684c83b59355a00d2da0c3 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a0071fd99e664bbab21080ae83a0003c : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f116359799524a919837d91249a91293 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/99819e2167d64360a44c291e58335dd2 : SUCCESS in 15m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79570fdc6fc149458a817ff5e26334ee : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41c41c2cff94492497a8a57aac02362b : SUCCESS in 1h 04m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1680d829c4a7470e89428c78272d66e9 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5bcfa6d81576c620f9a307387629149eb4fdacd7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac05dc8306ccf88d524e5ce06f69e18c43c6db54","revisions":{"ac05dc8306ccf88d524e5ce06f69e18c43c6db54":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:49:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/976774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/976774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/976774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/976774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/976774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/976774/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:39:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:39:48.000000000","tz":0},"subject":"update constraint for XStatic-roboto-fontface to new release 0.8.0.1","message":"update constraint for XStatic-roboto-fontface to new release 0.8.0.1\n\nmeta: version: 0.8.0.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I18795a43f74ea0e5601ffad8b911eaf63b453204\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac05dc8306ccf88d524e5ce06f69e18c43c6db54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac05dc8306ccf88d524e5ce06f69e18c43c6db54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
