)]}'
{"id":"openstack%2Frequirements~976775","triplet_id":"openstack%2Frequirements~master~I5b5ef93a93b4e76d3030641f9f456544b13044b7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b5ef93a93b4e76d3030641f9f456544b13044b7","subject":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.1","status":"ABANDONED","created":"2026-02-13 17:50:20.000000000","updated":"2026-02-16 16:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52b1634cd508b4c5921f9789329c227263b3932f","_number":976775,"virtual_id_number":976775,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:51:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"312be37cb5c2027c2c69f19455a2bd9285753089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:50:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3d8904dec06b432fb7a797ab80a45e4d5ee3415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:51:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b93416f3bda4201809bf66ee6ac3ac8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5920defe4d42dda1ac158ffcdd7dcb : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53db10d0bc664250acc96777d422e37e : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd121c0ffdfb4b019a9d7938e4fb7843 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83e5558836914acf87e47bc217e973ef : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bda4fefcabd64bd186109c9a262b6e84 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4712d7a57d3e413fb30e589dc4ad7e68 : SUCCESS in 3m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9ae6e23240054b3f9e6b899ae58f9b24 : SUCCESS in 4m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/acc24a06101a430689bb64e23191d8cd : FAILURE in 4m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1da43c03476b4b56a869efd7b8a4fc32 : FAILURE in 4m 49s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/af87a61eb9164f148aff31b2772660f0 : FAILURE in 6m 08s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/71557d969c5848e69f2c670520eeddd8 : SUCCESS in 6m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/39e5b8959bcd473aaae2ff93711bdbd8 : SUCCESS in 2m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1a1c872d5fc146cbb718568ad6043ac8 : SUCCESS in 5m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/caf3964def39406ea3ab6ded432dc067 : FAILURE in 4m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6fec9bc148d14a3796a451d3312c8954 : SUCCESS in 7m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/de9482dd905b49fb899304f9842a2d22 : SUCCESS in 6m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1c6c38ff3a7748178e9215d98ea4956c : SUCCESS in 8m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/aef1374894024dc0a0354f18d6bf29e1 : FAILURE in 2m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/810235a7841541d583406d3b6da596c7 : FAILURE in 5m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/392d1a725b01434ea29593a8ea789d66 : SUCCESS in 5m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5ff7150b506c4c2dae8fb5da277c8ad9 : SUCCESS in 12m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/84c7412f572e4fa2a717df4777cdd481 : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/caaf97f76bd94e328bb0188ec72848e2 : SUCCESS in 9m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f279c2fe6e1847568f3984aea205113e : SUCCESS in 5m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3d261788fbc34c9f838091565a8b4d36 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ff64b0b1ba51423da14409dfb864202d : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/00752474356b4697a01d9cb7fa6a452b : SUCCESS in 24m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5baa21f44dff4ea29b769b4d6c43bf34 : SUCCESS in 8m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cbc31b2feaec42cda5df114316794998 : SUCCESS in 9m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c1c445d77dc49b7b62dad628a6874a4 : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f9908e6ac9646858f2f2cadc1c8b918 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4607da8d1dc042cbbf79a4c1cfbca5f6 : SUCCESS in 26m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cbb57553fdc34125bd6418de02f2b814 : SUCCESS in 3m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/91943d0c018c482b91a760e0f6cd0337 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb24a699848a4193aeeea8c1327cfc40 : SUCCESS in 7m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/519dca595c7c4f05ae7bfd991b6a729a : SUCCESS in 7m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dd32e6275b004eee922f27ac3e12f445 : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47242e6a6ffb42758bd7c1821a16235d : SUCCESS in 32m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87f301128a664b99a39538b0240e8452 : SUCCESS in 1h 49m 07s","accounts_in_message":[],"_revision_number":1},{"id":"52b1634cd508b4c5921f9789329c227263b3932f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b22e373b452123b57b295b3c52126b33f5452721","revisions":{"b22e373b452123b57b295b3c52126b33f5452721":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:50:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/976775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/976775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/976775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/976775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/976775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/976775/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:19.000000000","tz":0},"subject":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.1","message":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.1\n\nmeta: version: 1.4.0.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5b5ef93a93b4e76d3030641f9f456544b13044b7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b22e373b452123b57b295b3c52126b33f5452721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b22e373b452123b57b295b3c52126b33f5452721"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
