)]}'
{"id":"openstack%2Frequirements~976776","triplet_id":"openstack%2Frequirements~master~I105b39ceb9b8d49b07fb79d28b45402d0a42c367","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I105b39ceb9b8d49b07fb79d28b45402d0a42c367","subject":"update constraint for XStatic-smart-table to new release 1.4.13.3","status":"ABANDONED","created":"2026-02-13 17:50:22.000000000","updated":"2026-02-16 16:22:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7524961ac9489f62720454057f782566a41fafe","_number":976776,"virtual_id_number":976776,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:18:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0278505af2a660fb15f6361bf0129813f7c14dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20b8930844a521fb015b044915813b858bdef078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b61bd300404c409d96f4d832a72f49c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b03983039342bc97fadee5c6df5568 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf04ab5a0304e2d899eb198da6b1da5 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f76f7578f4734a1c9958b7b7774fb37b : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42bd1bd778ae4723a48f4bf317cec79c : SUCCESS in 2m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1255c2d990914027b91fcf8634554983 : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ee0d8a218244fd4a376a8b21005ce96 : SUCCESS in 3m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/346bfe512ba9463d9f6389eceb264df2 : SUCCESS in 4m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9caa67eb034443bda1e500b9acfc71bf : FAILURE in 4m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9e1adbed52204b5582a560d622ffcc50 : FAILURE in 5m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f74734442e824e86914e319ef25585c8 : FAILURE in 3m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b9d27d6edcd94d3eb1692f0f5c0e471b : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/314d04819ee04337a1cbc1eb3adb7028 : SUCCESS in 6m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eab2463807474b9e95128181523167d3 : SUCCESS in 6m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4275a656ec7a40a5b37827650dc6d69a : FAILURE in 5m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8a364145e6354e2bacc6ff0775cc9348 : SUCCESS in 5m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d0ec4f4e6b48444abd6a461c8822f1e8 : SUCCESS in 3m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5a2c280e5d1f4b8f9eb169c56cf431e2 : SUCCESS in 8m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/311c57de25714e109836861f2e83f024 : FAILURE in 3m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d161c5327941439d8af47ceb14aa0984 : FAILURE in 4m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5ca72e4232fd440994d636039be925d4 : SUCCESS in 12m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/56759698dfcf497f9592d09799edd647 : SUCCESS in 12m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b3aea8219ae4743b3030fc8690f4977 : SUCCESS in 2m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a49241a6c36242b89552f21dda8bfaf7 : SUCCESS in 6m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1a88abed21f6431ab694c852c8baa7b9 : SUCCESS in 2m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/aaceaba230334fea889e7c43605a8af2 : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bdeba2d33c494bc3869d6c97f5ffaa90 : SUCCESS in 10m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f6d5346ff1f240979b4f3f04166c759d : SUCCESS in 26m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a73eb54e06b6487b9da1e27fc0c9bde1 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d6f13139f158434bac934b06e8187e69 : SUCCESS in 14m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e0aa7be9f8634508b4acb4eeb2a2ce9e : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23549350dc9c418a82e46b6baab717ac : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acc28dff63cb467790d1a003bf6f3bfd : SUCCESS in 35m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a4b50a5f7bfc4a25bd3817079676c705 : SUCCESS in 3m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f83b46967f1545aaaf4e4e904a774cdb : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/686eb85f8a1e4229a131239e5d357c3a : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd2fbac2f66c45ae80fbd3d7fe71df6c : SUCCESS in 19m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c1348d559e184565b27921c18e0382a3 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de90163e4b774ddaad77884eae388b79 : SUCCESS in 1h 06m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a4e36b100b1482292a1965aadb1d58b : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c7524961ac9489f62720454057f782566a41fafe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e939cde918b8586aff2dfee2e9727f748e1cddb9","revisions":{"e939cde918b8586aff2dfee2e9727f748e1cddb9":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:50:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/976776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/976776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/976776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/976776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/976776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/976776/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:18.000000000","tz":0},"subject":"update constraint for XStatic-smart-table to new release 1.4.13.3","message":"update constraint for XStatic-smart-table to new release 1.4.13.3\n\nmeta: version: 1.4.13.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I105b39ceb9b8d49b07fb79d28b45402d0a42c367\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e939cde918b8586aff2dfee2e9727f748e1cddb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e939cde918b8586aff2dfee2e9727f748e1cddb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
