)]}'
{"id":"openstack%2Frequirements~976777","triplet_id":"openstack%2Frequirements~master~Ie6c24fa37045bf615a0f089e8c53a3d678f32770","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie6c24fa37045bf615a0f089e8c53a3d678f32770","subject":"update constraint for XStatic-JQuery.quicksearch to new release 2.0.3.3","status":"ABANDONED","created":"2026-02-13 17:50:47.000000000","updated":"2026-02-16 16:22:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bacb2672251829b4bca8fee028205a40cf9c3504","_number":976777,"virtual_id_number":976777,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:49:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:49:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc869698056d6e05163bae98ed59af8007ea9a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:50:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32519c865432ab9355b40e67be3b205d87ecccaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:49:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fd53f18a9b54d2abeee8c890b907809\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356095a72fe043079f3ef771e74f8464 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3559fe6219eb4863922f8ccde19d4a78 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a686e412e367470ab89476253ad7d1d2 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0be4296d953149228a1807200f0f6eb5 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56c24422456a43d4bf3b5b86dd467b67 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb9a7e647a6d454fa13f497d4355a9bb : SUCCESS in 3m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a202288ffb94913a7716b04b2b7813b : SUCCESS in 5m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e95fcb3872894c5786a3d2ebfd6ee0b4 : FAILURE in 5m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb45573bcdc440029280380432480654 : FAILURE in 3m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6d0722aa12e4fc0912be19b06292962 : FAILURE in 3m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/96ca4219338b40edb72c73c22a705059 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6760b2a166ad4729ad13dd336a1b0f6f : SUCCESS in 7m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d798632c3c0f485b8b970298728e1195 : SUCCESS in 4m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d4615c8d83ea4597a10c50e30b1a14b7 : FAILURE in 5m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/40edd5f24ec4425d90c9ea018440b2b5 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/71bdd2069a694ec69c74749a8421532a : SUCCESS in 8m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e0cf55358c5c4e9f8ebcf2b8431f50f4 : SUCCESS in 7m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/25c55ad3795844a1b2df92db6ec3aef5 : FAILURE in 3m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85feadcbe0a74b96817759ad12766920 : FAILURE in 6m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa24b8105de44b748bc87e10639473c4 : SUCCESS in 9m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d0316f1ff052466f848b6bb7c7543b60 : SUCCESS in 12m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/633b8d426de6469d92ee271e375aefc5 : SUCCESS in 4m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c5853458346046bd8304b6791573dcca : SUCCESS in 10m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6280502831d24fe2848e40172c83efa9 : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9e6f316cde34401287dc16bf735ae2ec : SUCCESS in 8m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ebb8fdce843f4b3a8ebcca7a30f4b674 : SUCCESS in 10m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/26cf7e88caa446bc9ee66849972c2d03 : SUCCESS in 23m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17c55d85747740a3a21081025433c77e : SUCCESS in 10m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2c157b3ea5cd43aabb4f308c7be6eccb : SUCCESS in 14m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/83e8468ea0304ed4ad05e8090a586c68 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd3d36e65cd846e1ab3afce511ef70c6 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7574ce0039b4424896fc47aac68d968e : SUCCESS in 28m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/61d9223910ce480f8cbd5f6214562748 : SUCCESS in 8m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e3f9682c998d4f5597ee97bf82c72768 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d84f38f749ee4774bb2b9b1541ea209c : SUCCESS in 8m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8f06b5a8339f4dfe907be88e52c4813f : SUCCESS in 22m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/10dcaccd07e640eba81c8c6822aa73b6 : SUCCESS in 7m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12ae97d684ea4633a3c09ed8660b9c3c : SUCCESS in 56m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9ca0ef3352841a5afb41f30cd6566a2 : SUCCESS in 1h 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bacb2672251829b4bca8fee028205a40cf9c3504","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92e38e56447f91c66b1fdb4afb82cf4b998c8f21","revisions":{"92e38e56447f91c66b1fdb4afb82cf4b998c8f21":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:50:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/976777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/976777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/976777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/976777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/976777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/976777/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:40:46.000000000","tz":0},"subject":"update constraint for XStatic-JQuery.quicksearch to new release 2.0.3.3","message":"update constraint for XStatic-JQuery.quicksearch to new release 2.0.3.3\n\nmeta: version: 2.0.3.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie6c24fa37045bf615a0f089e8c53a3d678f32770\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e38e56447f91c66b1fdb4afb82cf4b998c8f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e38e56447f91c66b1fdb4afb82cf4b998c8f21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
