)]}'
{"id":"openstack%2Frequirements~976778","triplet_id":"openstack%2Frequirements~master~I868d52f5ad8ca68764fb3045392f356a31caf613","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I868d52f5ad8ca68764fb3045392f356a31caf613","subject":"update constraint for XStatic-FileSaver to new release 1.3.2.1","status":"ABANDONED","created":"2026-02-13 17:51:51.000000000","updated":"2026-02-16 16:22:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52f4e406fa8e2e6ee96c8d83e12ca3b80574fa96","_number":976778,"virtual_id_number":976778,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:23:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ae93f9239eeb59c86b51680bbe393ba8f6ba453","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9527944d94ca58b130487f4738107529289297bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:23:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e7ccf5a595421285623f364d7dfcf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40db02e1899a42c9ad2ec9389b6fa693 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2059a85077cf40bab053a1f44c0968d9 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b737586ba3a14319a2b11ef34ff51859 : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde75c2fc6714eff8a4cf082b31d3f38 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd20a5923925497daa4383d6c9ae5d97 : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/411ab3f6a0ed47ef85030c2e28eb57c2 : SUCCESS in 3m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3dc3dd80f17a43688412aa6245a602ff : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8dd06a8a96844d048423f397b0f0df1d : FAILURE in 7m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3cd345269c3c497ab0818167e6439912 : FAILURE in 2m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5751ef43d044470c9d7042138788df54 : FAILURE in 3m 47s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/79d782599c9241abbf84ee247703f6d5 : SUCCESS in 5m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ec8ba79c288945eb9ca1d71d96489e1d : SUCCESS in 7m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e3f86efeae854ea298d77031375049c6 : SUCCESS in 2m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/01c44543d1cf441db0a24afd056c1f92 : FAILURE in 5m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e4e40ce9def84c61be6f7993d7223024 : SUCCESS in 3m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/73c139f5cbab417ebb08fe6966ff0d75 : SUCCESS in 5m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/29b6f3784ea149eeb74a55da2ed656ba : SUCCESS in 9m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a6ba0a99aea242338dea5c6f6602f3f0 : FAILURE in 2m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1adc1b413b9471981f0299940c6b7bb : FAILURE in 8m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3396935bab4d4c6596ea6554c3c34430 : SUCCESS in 13m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/39306899b8be4ffdbb548f4dc3d00ed6 : SUCCESS in 14m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2286d6ec599149aca358b589c21b2641 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/103de300878c4c2998eaa6815e582c17 : SUCCESS in 12m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7d2d83b7a9ac425fba856badb89d1a88 : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1394dadbbd564e919f5db128d84480cc : SUCCESS in 6m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/874cc4ff3f5f4304a0a32d3decb17e84 : SUCCESS in 11m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8b0878657012401b840540e58dea1753 : SUCCESS in 31m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/31e139bb665d474eacdd5f723d92fefb : SUCCESS in 11m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/afad5c627380486c99745cc7e6940dea : SUCCESS in 15m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c7ff62ed06043d68c309659a36bb4a5 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc13a3581cc04055959cc0501e7783dd : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4f998158e7448a1a6c6e089948d4c85 : SUCCESS in 28m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0624434ad6eb4bf6be0def1722e91ffa : SUCCESS in 5m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/de3f064e4fdf4c2baa18fb57742a9fa5 : SUCCESS in 8m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62d10a2bf3ab421ca7c78cbe283956ee : SUCCESS in 7m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/13184e8f3d6d4a36b76c793ca267ab38 : SUCCESS in 18m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2f9e2632be344dfa9571357e8b3f266f : SUCCESS in 6m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e22b008d0d964fdda64dab0cae332945 : SUCCESS in 35m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff10d0b890534766a6aeefda426f3440 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"52f4e406fa8e2e6ee96c8d83e12ca3b80574fa96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abaa3386f9b651197d329170ebec817680c321c4","revisions":{"abaa3386f9b651197d329170ebec817680c321c4":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:51:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/976778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/976778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/976778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/976778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/976778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/976778/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:41:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:41:49.000000000","tz":0},"subject":"update constraint for XStatic-FileSaver to new release 1.3.2.1","message":"update constraint for XStatic-FileSaver to new release 1.3.2.1\n\nmeta: version: 1.3.2.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I868d52f5ad8ca68764fb3045392f356a31caf613\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abaa3386f9b651197d329170ebec817680c321c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abaa3386f9b651197d329170ebec817680c321c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
