)]}'
{"id":"openstack%2Frequirements~976779","triplet_id":"openstack%2Frequirements~master~I1f49e106e6412b93b00fe9fa5fa0e9640f276643","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1f49e106e6412b93b00fe9fa5fa0e9640f276643","subject":"update constraint for XStatic-Rickshaw to new release 1.5.1.3","status":"ABANDONED","created":"2026-02-13 17:52:25.000000000","updated":"2026-02-16 16:22:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64405cf58076b2197f4f94423b2843f3f81daba5","_number":976779,"virtual_id_number":976779,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:29:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af3657a7c47e0391ce9c54b83120f49683c33f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:52:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcc67d0655655fb1a2704e5bb3310badf3fa7d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5e66cef684f4528ae226b8c047d3fbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8e94a035e4403caf63372c5d664226 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad0e5deb68549f4ab384da0ad79268e : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f85f0a4b14c40ea9ebce8d18b855c81 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/499ab603c5884fc59dee6c6401a84166 : SUCCESS in 2m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/882f36a4294b4a939f2b27f1ffda606e : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9e48d3425e64d8b807557d28e1688cd : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2019ba67538841528f5f8758d025e85b : SUCCESS in 3m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bc8b1909996143aa85b1f65a54060a28 : FAILURE in 4m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/287a6d9129944acda73b7532a763949e : FAILURE in 8m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/808a354be9134a998aff3dbda2dfae92 : FAILURE in 4m 39s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/03e2e08f70cb4828bd48309e88dca35d : SUCCESS in 7m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4e8323bea98b4750bb03975bcc187ee4 : SUCCESS in 5m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2892815504194f0eb912bb13aed6b5a9 : SUCCESS in 4m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/eea7c9bc55d9492aba9ebf324f79d62d : FAILURE in 5m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/173acf5fa4f34622841042e9fdc60ef4 : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8e50f7a11cdf4d9ba8bd583c3a1fd1e6 : SUCCESS in 4m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7d3a235d79a946f2a3f8f24dbbfb84ca : SUCCESS in 8m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/beff3e833a1f4bdebd0cc5ce8a59b37e : FAILURE in 2m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebd8fbbe164d4b0ba6832a4f7c375106 : FAILURE in 4m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bf82f1581f044fdda31dc9550ac89476 : SUCCESS in 11m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f4914c6246cd4b9494c59aaf1f445b0e : SUCCESS in 8m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aed56da9eb164d7eb5aa42a27b72a6f7 : SUCCESS in 2m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/281a960adcf64587a325e6963b7956fb : SUCCESS in 9m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d7eba77e91ac499793d29464e78d899b : SUCCESS in 5m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d6cc98c10b7842e5a0261fe042ee9e14 : SUCCESS in 5m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9b8e753da364466480f49381ed10fc61 : SUCCESS in 12m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/179601afcbca4395942b27040e17c9d3 : SUCCESS in 25m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b2c9d0c840954501bd19e03f1abe25ad : SUCCESS in 4m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a908a3f644f54dc88a521750bb360066 : SUCCESS in 15m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0302cf27c876456a8f06137bc53e44f6 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/704dc0d2be934926b9e66b8fa936581a : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/204ae7db37374414a6e0366fecb3af1f : SUCCESS in 29m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/190bb146239a44e1b830d7887b530397 : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d0d1343a9c4840bb972d307c43fc1d54 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3d3bc0ab7b346938704550f19a4d46d : SUCCESS in 8m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9d495d1863e1447a89e3f786b0ccb5d3 : SUCCESS in 8m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1fcc0fe6dad54deaacef23cf315a6c16 : SUCCESS in 6m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32d3c63ecb5b4166b7a39e5e99ddacdf : SUCCESS in 59m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0859787ca6b64f2ea9d7281dde3a9e76 : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"64405cf58076b2197f4f94423b2843f3f81daba5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4c8deadae92af756a0e92ee845a5d35b96e2272","revisions":{"a4c8deadae92af756a0e92ee845a5d35b96e2272":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:52:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/976779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/976779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/976779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/976779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/976779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/976779/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:42:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:42:24.000000000","tz":0},"subject":"update constraint for XStatic-Rickshaw to new release 1.5.1.3","message":"update constraint for XStatic-Rickshaw to new release 1.5.1.3\n\nmeta: version: 1.5.1.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1f49e106e6412b93b00fe9fa5fa0e9640f276643\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4c8deadae92af756a0e92ee845a5d35b96e2272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4c8deadae92af756a0e92ee845a5d35b96e2272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
