)]}'
{"id":"openstack%2Frequirements~976780","triplet_id":"openstack%2Frequirements~master~Iad24370db0bdfaa15f209168c78fa0577f18fa43","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iad24370db0bdfaa15f209168c78fa0577f18fa43","subject":"update constraint for XStatic-Angular-Bootstrap to new release 2.5.0.1","status":"ABANDONED","created":"2026-02-13 17:53:01.000000000","updated":"2026-02-16 16:22:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2899e4796ff7bb965a6a61b288bf8f8b4b62310e","_number":976780,"virtual_id_number":976780,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:16:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aef2b96e40db12e0be1e763c3793c1a5ccddb116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0fe573c15ff06e4a8a63729606512d6179902ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:16:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0cbf1f0dfa343b89a2ca5c701c3e26e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96c1c0d4574477fb6fc02587e59a278 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3bd24eff340496c9caf9a79a61b188d : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43f1cb96524446b38fea5e36d6e775ad : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b0e75b4fa0a4f56b507637b5d41dee5 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ab7f4e3b62445e4926972a1390948c9 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3cc41692059d4fe29f2deb418cdf1a5e : SUCCESS in 1m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4b706ff7dfe047e8913d5985b503508d : SUCCESS in 7m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4d819fec41024040a68cde56cbf58cac : FAILURE in 2m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/039918e31cf3453797793ab8b1ca9ff2 : FAILURE in 3m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dfd8e41c70d24e29b617b3ddc8ca6883 : FAILURE in 3m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4194b1edb5394ddd80719901630580db : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/70f3bc07142b4ef5819d4017b947d940 : SUCCESS in 6m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ef53dd336a804bfbba8bed4a8d81b3bb : SUCCESS in 4m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/579b6621c2c94dd1b3e685f56b94f763 : FAILURE in 6m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c35301066f304c40af861373b13d3f05 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b6cdab0d837d4fe2904e1cb8c180395e : SUCCESS in 6m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c05df9a6e426499da3aec7146871e82e : SUCCESS in 8m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e0d2765731164c2d8c9b267c55a473f0 : FAILURE in 2m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62118af004c943858a47fa195be2c1f6 : FAILURE in 5m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9665d383c4b4408a954eda9089daa7b3 : SUCCESS in 8m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1abdb4d21a914422ac0f252189c6719d : SUCCESS in 9m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e6333d6382ec4cf884e039d6c5f8e8e4 : SUCCESS in 3m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/689f907b077d477bbfbfabc2fd2d6d0d : SUCCESS in 12m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ee1b29d3cae94dc992f912a0eb33f975 : SUCCESS in 2m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1a7f20141ef34c61a281ef1fa770b4a1 : SUCCESS in 7m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/83f47fe399224bc384f27e6f915a924f : SUCCESS in 11m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/40b3b3e247bb41418495fbf93ab7af98 : SUCCESS in 20m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bfc70ad6453d4c38ad650d887409bc6f : SUCCESS in 9m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/33da45630a93409597f5b1acc5b6707f : SUCCESS in 14m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8520825832c64d2d9f02237084eecaab : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69ba0f45300a4fc894b04bd2ecfd4586 : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d7e9afbccbd408ea611fe1ac74532da : SUCCESS in 27m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7e897191a1ef44bc9127b47ca10ca34d : SUCCESS in 7m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bba36e42614b49afa12031b426f08c7e : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/888cfb06ae254be290351870af6c12f2 : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/778e9a58e5484b3da2fc2f910ce70daf : SUCCESS in 19m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a2bc90cbc9f843b6a111f71981f25236 : SUCCESS in 6m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/475941418daa466bb90bcfdad03fbc0c : SUCCESS in 1h 10m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0be908cd42884b57ad3e0fe982b8c087 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2899e4796ff7bb965a6a61b288bf8f8b4b62310e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df4e61f4be9120aa4e980da28c1ddbd33a0070d7","revisions":{"df4e61f4be9120aa4e980da28c1ddbd33a0070d7":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:53:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/976780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/976780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/976780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/976780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/976780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/976780/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:43:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:43:01.000000000","tz":0},"subject":"update constraint for XStatic-Angular-Bootstrap to new release 2.5.0.1","message":"update constraint for XStatic-Angular-Bootstrap to new release 2.5.0.1\n\nmeta: version: 2.5.0.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iad24370db0bdfaa15f209168c78fa0577f18fa43\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df4e61f4be9120aa4e980da28c1ddbd33a0070d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df4e61f4be9120aa4e980da28c1ddbd33a0070d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
