)]}'
{"id":"openstack%2Frequirements~976782","triplet_id":"openstack%2Frequirements~master~Ia76a0541bd528035060adfb63e17e12b4a8eb46b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia76a0541bd528035060adfb63e17e12b4a8eb46b","subject":"update constraint for XStatic-JSEncrypt to new release 2.3.1.2","status":"ABANDONED","created":"2026-02-13 17:55:13.000000000","updated":"2026-02-16 16:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4cf577c33ff1330607f531e9f00b84970c95dbec","_number":976782,"virtual_id_number":976782,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:03:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:03:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b74f5d352bec7b4c9e11f8884f59970a29324868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21e90dd048dbfa48efc2ac687662757e86a0760b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:03:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f402ccd1244e4d028df9d24221b4c8d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd0d7ad1bae46d6ba2f14d95cecdff4 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa9595cdc87447397c8fd1c0d0f8ad0 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cee1aefd1b3402f9e9ea4f27eea0eb1 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43874de584e7493792f282cff565340a : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3641ec3df3d427e890160ade8af1e2a : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9598064e3a5f4012a5d639c436ad1643 : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e7d2b79abfbc494d98eb388696b5a72b : SUCCESS in 5m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/df17d98507da433a9d1e260d0b825be9 : FAILURE in 4m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7e6fb5b1b38d4067af904d12863916c4 : FAILURE in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/65aa25e2666747c188908427c7daf84f : FAILURE in 5m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/13f8125ff0fb41828e7c74e14f004c05 : SUCCESS in 5m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a03ec4461afb41ab8483f8bb3298ea74 : SUCCESS in 4m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dff40a3efbd240bc98affcdf4ebba9b4 : SUCCESS in 5m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a4e6c11d82e446a2b1d2ad586321e8d3 : FAILURE in 5m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aa6b3b06c8044e359ada2b9a3d93af50 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/68e9d942f01c45408e870ce082358a55 : SUCCESS in 6m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/775a8c21958c443fbf0de32d520591f7 : SUCCESS in 8m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/729f6f4d5a624ea1a20e83b6a754aec7 : FAILURE in 2m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/821ee6a324474369bbd17f77c7eb0aed : FAILURE in 4m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c944cad493a44df7aae9823c63e45be0 : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/79457c70648b44ffa82cea0c3d51b519 : SUCCESS in 13m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c5689933c05c408a87c840ef3485f0b4 : SUCCESS in 4m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/47b807fedc2940f7ad4de60fc1613bdc : SUCCESS in 11m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f1c3ff4705854d17a17835aa6d26297d : SUCCESS in 5m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3bbfecc6af2e411f9903379e1e10770a : SUCCESS in 6m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/660ca5074520464cb65773b8c27bf70e : SUCCESS in 13m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/92d68fd1f6b548ceb8dd601f8c179287 : SUCCESS in 27m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e204cc8796ac4702bbaef806a2173774 : SUCCESS in 8m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cc15f3f07fde46a0ae65008e2c9bb96a : SUCCESS in 15m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/da60cbad83214134860e3d8c15dc2260 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42558a5746b34c4191075fbd4a278b4b : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1b98b8045414c7d8f783841bfd291ce : SUCCESS in 27m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a903c760dc51485fb58a849045583ca4 : SUCCESS in 5m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3b38c6761d0043f3bf090f24c68026d3 : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6bd832c7e8f497081350e3a5082c7a9 : SUCCESS in 7m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/184a865346ce4bf59042c5d387a4ee66 : SUCCESS in 19m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a4c872480595417aa2c16fda0bef7be1 : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d542feeeebb4829b9082c2568ef1869 : SUCCESS in 37m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b09f3f1724ac4390a8bbc384f9268299 : SUCCESS in 1h 44m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4cf577c33ff1330607f531e9f00b84970c95dbec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99ba016e2420b20970664663cf8ac2183dcf7acc","revisions":{"99ba016e2420b20970664663cf8ac2183dcf7acc":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:55:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/976782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/976782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/976782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/976782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/976782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/976782/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:45:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:45:12.000000000","tz":0},"subject":"update constraint for XStatic-JSEncrypt to new release 2.3.1.2","message":"update constraint for XStatic-JSEncrypt to new release 2.3.1.2\n\nmeta: version: 2.3.1.2\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia76a0541bd528035060adfb63e17e12b4a8eb46b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99ba016e2420b20970664663cf8ac2183dcf7acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99ba016e2420b20970664663cf8ac2183dcf7acc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
