)]}'
{"id":"openstack%2Frequirements~976783","triplet_id":"openstack%2Frequirements~master~Ie57db386be9b1bc3252c4776b7dcf26ee8926cb1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie57db386be9b1bc3252c4776b7dcf26ee8926cb1","subject":"update constraint for XStatic-Hogan to new release 2.0.0.5","status":"ABANDONED","created":"2026-02-13 17:55:56.000000000","updated":"2026-02-16 16:22:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13943257f402800d8a90af994fb34985f5997348","_number":976783,"virtual_id_number":976783,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:48:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43243a5ed44c13fe9eed27b3afa8227a67bfbd77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76920b29349d87d50ef96f546981ce0f48e81f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ef3c6efdb847768774974e5bc901b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5a580103ac474ca77d768d5592d0d4 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67217e25a2f9498cb74256ca92fe6f64 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/280029d6c24948dfb8c2e850143a4bdf : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b054427d41e1454b90185ecf4c73ec7d : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38f65409908c4d679cc5aa3efb279208 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8f078371576411a8f842b24b758ae31 : SUCCESS in 4m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b16558a1d26e421797910662e8a60824 : SUCCESS in 5m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/27ef379032aa4a3bb09a14e44e3800af : FAILURE in 3m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1055efe51ccc46bb8bc4f65491150195 : FAILURE in 7m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0acce1cb2641469daf9ef482a37662a7 : FAILURE in 3m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/039ff488eaf14223aa27d3e656882a54 : SUCCESS in 6m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ad67b9fe4c8b40e49553a5bc6a3eb713 : SUCCESS in 6m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9a98d978286f49b89b2dac4688b1c96b : SUCCESS in 5m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dc633bc6d372413cb36c8c127e61d7ce : FAILURE in 5m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3a503c6fd2dd486490819ce3804a05f2 : SUCCESS in 5m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3c0aed3d549540b39d9d8dab9b351687 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/27290c868ea24c0085deb783cd3299ca : SUCCESS in 8m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f2fb706103874b038c7697224e45bf9f : FAILURE in 3m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10a817acabb64729ae502abf4c015c1f : FAILURE in 5m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b923ec85bc0947e0a420ccd8ab0d9ff0 : SUCCESS in 6m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9c1e9c7427da4808aa5703fd82700cbf : SUCCESS in 12m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/659bb9d43eb64110acf78a0f2cedf71a : SUCCESS in 2m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/54d2673cc0c34bc29c53a92b1878f77f : SUCCESS in 10m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f1afcb6663e14006ad13ecfa0507914c : SUCCESS in 3m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/99007080044d450d8790724c91d43d3c : SUCCESS in 5m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/483e1d8362a24b8cb0f4e38e0c0e9da7 : SUCCESS in 11m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8433b871605e468bb5bcd8b2c744ca2d : SUCCESS in 27m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f5bb95b9a9a8422a9be57058d8a97b10 : SUCCESS in 4m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cf5b9b4bbcfe44119f28e1541b954ce6 : SUCCESS in 15m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9e4fc271aee2421597673543c80b12ae : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a69a861446914843aef27b1fdd8e04f6 : SUCCESS in 2m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53fa371e2c2946b9a18b2b37898c0eed : TIMED_OUT in 41m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3e708de63e294c4a88c56838c36a0f26 : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2d42499eccbd42d4b6e87588bc575674 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ba271a4cf6147398116b7973ac25a04 : SUCCESS in 3m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/78f1ea27943c47f6bbc68218a604f7d9 : SUCCESS in 18m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/071b11a44cde4c7caccd26fed6954fd2 : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5814dd24a4bd446c934b2e1d8ffb7bb5 : SUCCESS in 37m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/492e620c1ffe4e29abb4436f6a8f016f : FAILURE in 1h 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"13943257f402800d8a90af994fb34985f5997348","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3b2f95b87cb7e1a55a81d1aad7b15e71eebc1be","revisions":{"b3b2f95b87cb7e1a55a81d1aad7b15e71eebc1be":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:55:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/976783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/976783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/976783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/976783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/976783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/976783/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:45:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:45:54.000000000","tz":0},"subject":"update constraint for XStatic-Hogan to new release 2.0.0.5","message":"update constraint for XStatic-Hogan to new release 2.0.0.5\n\nmeta: version: 2.0.0.5\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie57db386be9b1bc3252c4776b7dcf26ee8926cb1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3b2f95b87cb7e1a55a81d1aad7b15e71eebc1be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3b2f95b87cb7e1a55a81d1aad7b15e71eebc1be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
