)]}'
{"id":"openstack%2Frequirements~976784","triplet_id":"openstack%2Frequirements~master~Ib3959636b2647855d95c4086004f5cc501245c00","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib3959636b2647855d95c4086004f5cc501245c00","subject":"update constraint for XStatic-Jasmine to new release 2.4.1.3","status":"ABANDONED","created":"2026-02-13 17:56:05.000000000","updated":"2026-02-16 16:22:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1694f26bf12ba8419dee45c49e8dbf52fbbb2018","_number":976784,"virtual_id_number":976784,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:20:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85721bcec91bc15573cdfc0bf6320e02fff5f13e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c761f84256891c04964655310a6eaa810f0d3381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:20:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b872d0bec27744229823c3a5448864e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f22f40d141a4beead586dde24505c22 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f8aa8fc6cc40018f843065f2fc2fd0 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/536112921ddc4d2b9ced047ac5c30ca3 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27e0d7db7d934806a99d12a010a8d6f2 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd8d8c680ead422d96d959ac439186a1 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b607b57e219f493e9d509391ae7ff6ae : SUCCESS in 3m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f2a9de202f6f40c19fb7c1b29cf35409 : SUCCESS in 4m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fb81633f6c434a56a01e5d406918cd6b : FAILURE in 4m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/827d1c1d57de47d18f98e2a71e28f71b : FAILURE in 4m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/22c909d0ae3545f8b0a9ae55a80a4cff : FAILURE in 5m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b8ad975fed47415aa51f1cfc116cc235 : SUCCESS in 5m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5a968485b4f740159beb77160b64d7b4 : SUCCESS in 6m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/439c8240a79b4dcf98716c157803c78d : SUCCESS in 4m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f323a57cf2124772ba25a315136f6ec7 : FAILURE in 5m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b3a4ff9dd3384a03906e478dfff9e243 : SUCCESS in 6m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a1e7cb1f84b34100be63c5519e13a789 : SUCCESS in 8m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cfadf1242e0149e994e1cc1b83a10a1e : SUCCESS in 7m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d9a42dd5da9e4c8c95fe13e758295fc1 : FAILURE in 3m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/017fd80f09234492867f7f1a72677218 : FAILURE in 5m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c5c8a116944944efbb0e50b77fcfc4f4 : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/70b60ce6dbcf4668a87c3c452605eb00 : SUCCESS in 10m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7be2a12a2dc948d2b69ef60b9687fa9b : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/01fe701e26e3415b9eab696f54c6b8b1 : SUCCESS in 10m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/07243d76c7d04dac960bb0bd116d7506 : SUCCESS in 8m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/86fef0b57be54c33bf45884d03c44686 : SUCCESS in 6m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c37a3a2b943e4e4bab49905ca8298206 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/18138448c781402caa8b0f29360eaaf0 : SUCCESS in 41m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01a7708d401641e78cf53569f2552e49 : SUCCESS in 8m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/beb907f0005f4994a6f2a8f9398cd832 : SUCCESS in 14m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/17595f02228d44bc8291514f7937cad1 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/091920d152be4c2aa696a28b40eb41c8 : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf16bea171154f4f916a73a7884ff677 : SUCCESS in 27m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e6815173756c4813a6369b46fb8c750e : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fad3837dde3647ecbfb8f89a54dbafef : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3d9163c9bfc48fdabe6eb3466f341b9 : SUCCESS in 8m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cb5378ea0c964475b574878909184351 : SUCCESS in 30m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7a65579c932e4d37a578b563a0be8ff5 : SUCCESS in 7m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c06e214bb90f4962a4869396456e494d : SUCCESS in 1h 06m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5ad629a7b094b4f98a48fc4a47c3118 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1694f26bf12ba8419dee45c49e8dbf52fbbb2018","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fccfb3b7fa03b81f8ee18a190409019ad9ecb7cd","revisions":{"fccfb3b7fa03b81f8ee18a190409019ad9ecb7cd":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:56:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/976784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/976784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/976784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/976784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/976784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/976784/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:03.000000000","tz":0},"subject":"update constraint for XStatic-Jasmine to new release 2.4.1.3","message":"update constraint for XStatic-Jasmine to new release 2.4.1.3\n\nmeta: version: 2.4.1.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib3959636b2647855d95c4086004f5cc501245c00\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fccfb3b7fa03b81f8ee18a190409019ad9ecb7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fccfb3b7fa03b81f8ee18a190409019ad9ecb7cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
