)]}'
{"id":"openstack%2Frequirements~976785","triplet_id":"openstack%2Frequirements~master~Ib125ab7389aab89e05f86aeeebc7f4d3e2f61ce8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib125ab7389aab89e05f86aeeebc7f4d3e2f61ce8","subject":"update constraint for XStatic-Angular to new release 1.8.2.3","status":"ABANDONED","created":"2026-02-13 17:56:35.000000000","updated":"2026-02-16 16:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33dd03cc676898c05240f89b8e8151a7c38ca39b","_number":976785,"virtual_id_number":976785,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:59:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:59:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"196f57b55004e7acabd8217a33346a6ee22ed6cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e9d565f1e2198662576c3ae6f3ece8d4494a4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:59:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8569060fb90e433fb55244b13382b25a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097605f8b12f427496939786f09a9bc8 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33767f0f67564aefa8dd8f8809e24b5c : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c40a126273954c8d8cede99eaca653a9 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d36da0178c640bfa767cba0d5ef05a6 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e5c7fd068c84c74825cd09bd21ecb6f : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed328faac9e24efa92307296757f0bfc : SUCCESS in 4m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/67ca1820f3fa4cc8b6d1c55113122916 : SUCCESS in 5m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8de8630e7de345d686d68a1af567aa56 : FAILURE in 4m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/43257617085b4a32879ed6b6c7e5069c : FAILURE in 5m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6d6aacac4594be88a305cac86894dfd : FAILURE in 6m 20s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/efac28d348104e1082ad72ca130b9aa6 : SUCCESS in 5m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fe9c17e1453043d0a64d95fd88fb5de8 : SUCCESS in 5m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4e72b7fef6984bfc851de869c53bbbb3 : SUCCESS in 5m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ed5c8a583950428aa252ecfc4cfd643b : FAILURE in 6m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cad0de5aff9847718158d4cca2466401 : SUCCESS in 5m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/be63635b602b4aa99e0d2618f25a76fd : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a2d426cdaae345698edef0e3231cb4b6 : SUCCESS in 8m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/28c6a8a6cb6340579290f2d1511ab64b : FAILURE in 3m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/041867fc0e434688b9c8f365c881c174 : FAILURE in 6m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/987781bb25ce4f52bc14d4cb8e0893a9 : SUCCESS in 8m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/74ad6b543bbd417091b7549bae7e0159 : SUCCESS in 11m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f30f5c4bb6f84e979de9e90ef442ec92 : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3039cd4a5b3a461ba610ce49bc0e55d4 : SUCCESS in 12m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/968737efbabb4ef38e90c40b859093cc : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6e7cb7656f374d5c9cad29ef2d87f23b : SUCCESS in 6m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0bc464cd1ef647eda82f0f378c682480 : SUCCESS in 10m 33s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c5c3154c3616416db47d6f177950b621 : SUCCESS in 22m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cbf53ff29048456689541f3b1c930684 : SUCCESS in 9m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c90f6533c0734194afe28bee1dee425e : SUCCESS in 9m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/34051ec06a1740f49ecb37e2c78262f7 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de66f472d01146d78a9392190c9398bf : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8261b230af984a8d893c999c6ea7da37 : SUCCESS in 25m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0bc217d77d47453cb77b125dbef07e49 : SUCCESS in 5m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4f955b367b6f48afa53d41ea86f0fa32 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/163d1dd676e643e7a2c3e9ed45887c89 : SUCCESS in 7m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6000d6b3646f42d8b10f337bf07b21e5 : SUCCESS in 8m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/123c0ad3ee3a416995a7b3819efca493 : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fa4e6b147064d67acbc7d6a0fb6ec89 : SUCCESS in 1h 10m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a12866f30f547ad89a4a0a18b4e1f4f : SUCCESS in 1h 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"33dd03cc676898c05240f89b8e8151a7c38ca39b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2d0454f3a51207b983e838fdb1ceb262ece57e4","revisions":{"b2d0454f3a51207b983e838fdb1ceb262ece57e4":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:56:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/976785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/976785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/976785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/976785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/976785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/976785/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:33.000000000","tz":0},"subject":"update constraint for XStatic-Angular to new release 1.8.2.3","message":"update constraint for XStatic-Angular to new release 1.8.2.3\n\nmeta: version: 1.8.2.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib125ab7389aab89e05f86aeeebc7f4d3e2f61ce8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2d0454f3a51207b983e838fdb1ceb262ece57e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2d0454f3a51207b983e838fdb1ceb262ece57e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
