)]}'
{"id":"openstack%2Frequirements~976786","triplet_id":"openstack%2Frequirements~master~Ib8a22a5d37860d8037138fd37d795e650443ada1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib8a22a5d37860d8037138fd37d795e650443ada1","subject":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.2","status":"ABANDONED","created":"2026-02-13 17:56:41.000000000","updated":"2026-02-16 16:23:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"05f4abd76a0c97253b097bfe26e0f1446fe563b1","_number":976786,"virtual_id_number":976786,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:11:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:11:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d30ec7baa0d748568c8f043ad01375f9f7d74d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:56:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20835d1db1a33491d5540cbcb9c2705828201d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:11:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4d9fdf88d27471ba66f205c96642386\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a65ea46d9c4b1abc5365b00ed753cf : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b911b6101e46459a87a06b40d64f710c : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00ecfbfce3a641318fb90f6079a99fc8 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66f39785c702465d9edea19ead2cd8ba : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/867b389a05b54140a01be1b32fad97c1 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e22de472bd5941c7a754294ee4625008 : SUCCESS in 3m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a4d6d6b2af0c49edb2402c0264de4e8d : SUCCESS in 5m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f38d682591543d59065d9d1840d0909 : FAILURE in 5m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9fa4498efba848b0993b9908cf1375f3 : FAILURE in 6m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b3bff7cfdd8f4fdaaeed18afc0e82f4f : FAILURE in 8m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6dc6e088270b444e9ce24947cd1bd4d4 : SUCCESS in 7m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/187a0fb115d84c0284067b62fe026b4b : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d19eef31249d4739aed188bc6423656f : SUCCESS in 5m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5d7e4ddc31f24ca89cddfff6a516100a : FAILURE in 5m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/997f85f3bd3a4705b78a019173df8da2 : SUCCESS in 6m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6befb13f4c29446191c25949854f34e7 : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ad58eb2615db4a4dbf330e6400a11d37 : SUCCESS in 8m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8a950d33a439476b928db9c9e8c02744 : FAILURE in 2m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39ceb273efb6426aaae7b8509c2a6866 : FAILURE in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8d7b5fbc60a5409f9a8e8ada8bf57d89 : SUCCESS in 8m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f10b661843b349de9b92d40827c4017a : SUCCESS in 7m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9551bd37a3b149b69cafd4dd01aceef6 : SUCCESS in 3m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b6d41169e27b4afaab12dc0ce8e4309c : SUCCESS in 7m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9865e16caca847d8b91472e4c5617382 : SUCCESS in 6m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7e2bce57ced8483182deb5b3dc11e7d6 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9056c1bb613847e2a1d28a57cebf928a : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/47bfada26ed94013a2305e4a508a0447 : SUCCESS in 23m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4d24034298624de0b5a2cbaa916ada2c : SUCCESS in 5m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4f8bf8f6f40243dba38731e26079278a : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ee0f7a19bfd8424b867c419f56f06e77 : SUCCESS in 7m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30afdd84e6df44ed938e49f4fb2e23d0 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e385b93bf0e4adf95a15ce09b80d3e4 : SUCCESS in 25m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a66e79dc16fb4fb185bbcbafb9e93d67 : SUCCESS in 7m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2acb69d90d3b4ef696f2a56791712037 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82c5854053a340c2bebc62c295e99ef8 : SUCCESS in 9m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/50e5b515182847f2b8cd008ce784bc3a : SUCCESS in 16m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1bfeb4f888df4715ac5fe85fdb354791 : SUCCESS in 6m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6dcc44716414be1a0251dd044197689 : SUCCESS in 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb6f9dec2f6e4a4e8bf6dd5f4b4ccb79 : SUCCESS in 1h 50m 08s","accounts_in_message":[],"_revision_number":1},{"id":"05f4abd76a0c97253b097bfe26e0f1446fe563b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:23:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6529cbc7dbda5fef8731f480d01f42acc739353d","revisions":{"6529cbc7dbda5fef8731f480d01f42acc739353d":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:56:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/976786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/976786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/976786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/976786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/976786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/976786/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:39.000000000","tz":0},"subject":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.2","message":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.2\n\nmeta: version: 12.2.13.2\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib8a22a5d37860d8037138fd37d795e650443ada1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6529cbc7dbda5fef8731f480d01f42acc739353d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6529cbc7dbda5fef8731f480d01f42acc739353d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
