)]}'
{"id":"openstack%2Frequirements~976788","triplet_id":"openstack%2Frequirements~master~Ie1ab15055b863c93af0c7d47b31badbbdd6393cd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie1ab15055b863c93af0c7d47b31badbbdd6393cd","subject":"update constraint for XStatic-Bootstrap-SCSS to new release 3.4.1.1","status":"ABANDONED","created":"2026-02-13 17:56:57.000000000","updated":"2026-02-16 16:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e3357830eb6504e91885f5fb77e8087ed4ef62d","_number":976788,"virtual_id_number":976788,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:08:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67e88ce128d838b817f10195f0fd6aeb3d1a48a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:56:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"def1a7375f5f4d63ac6ec0d5eb979500025078b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c35e3d6035c74143a312d70e3a22e208\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6487b4ad5f2b46c8b8225b596ffa9e74 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c88748768043c69146a14183bd7700 : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f1fa0c5d19a49f7bae25108384c692a : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4b00c6bc1144be4809771a0c9c7247c : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fa4535f237545b4b8770c7c1ce7ddce : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c8b4cd58049437bb5120a5a5a80ebe4 : SUCCESS in 6m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4950af93c6a0488e93a86dd6040fa0ae : SUCCESS in 6m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6b72ca174ee5497ebdb8041afdd24f95 : FAILURE in 4m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/86e9a9ced2694fe19818b6ef85cdf844 : FAILURE in 7m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1cdbe9fe4f914e5e84d0051b03ea55dc : FAILURE in 5m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9fd8058ef09f4b3d83fb6e9261ac446d : SUCCESS in 7m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/000914c470024c74805ed276123486c1 : SUCCESS in 8m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5362e3ef6e9b441b8f29f066a0b284d9 : SUCCESS in 3m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c1aa78aac2c24ee6a225979549c0e1e5 : FAILURE in 3m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0a07b81f3864406fb00b71d6a2b66fc8 : SUCCESS in 8m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d96b6b9399de4308900168c9d97a0f9a : SUCCESS in 6m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f04b884d8338412d82ab8906148c563b : SUCCESS in 11m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8d0fe833d2e245c5b7e0d52c6b372907 : FAILURE in 2m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/559faf096f2a4e4190dc10fde76df041 : FAILURE in 6m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/63c9e733f2ad40898984a974caf6ed78 : SUCCESS in 9m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ead3c590de684312b64559f3b41117fc : SUCCESS in 9m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c3025bf0d93c48f2805ad2a69b018d90 : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/58eebf5f2d4e4f3d86bd31629d08aee0 : SUCCESS in 12m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/06c4e7666d0f48e08379b11131ee5a95 : SUCCESS in 7m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1d224c1bd4a94212b398d9c3f6f387bd : SUCCESS in 3m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ed693661bd644b74a98bc5cc7433071f : SUCCESS in 11m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6a8a3538504d456b83a256fe73ee7793 : SUCCESS in 29m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d377a608bbf3483985ae6b80208be356 : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2ef73e31377e4498a95814c8b3f5836d : SUCCESS in 12m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/045fda8b30bf4bea84dfcf36eed4f517 : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47718e54416747f9a01aefb04786cb43 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/823ead988e9c4c73a33f9e55aa7960de : SUCCESS in 28m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b175eb566270487db0ae164b2642a11c : SUCCESS in 8m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/49c9018d718d434abe4a0c0ddb777729 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af253d8e027e44b7826c060398048a63 : SUCCESS in 4m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ae7a5364f29942348d0874a1d866f0f1 : SUCCESS in 14m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e6ab34a621ec46b984cd47de967d04c1 : SUCCESS in 7m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0559af8b40604598a360aa7a44836fc3 : SUCCESS in 33m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1219d9b8f2d043c1a243148511a9bf07 : SUCCESS in 1h 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8e3357830eb6504e91885f5fb77e8087ed4ef62d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6618f335bd26a489f5d191cf36a4d480bab5ddc","revisions":{"d6618f335bd26a489f5d191cf36a4d480bab5ddc":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:56:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/976788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/976788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/976788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/976788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/976788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/976788/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:46:56.000000000","tz":0},"subject":"update constraint for XStatic-Bootstrap-SCSS to new release 3.4.1.1","message":"update constraint for XStatic-Bootstrap-SCSS to new release 3.4.1.1\n\nmeta: version: 3.4.1.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie1ab15055b863c93af0c7d47b31badbbdd6393cd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6618f335bd26a489f5d191cf36a4d480bab5ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6618f335bd26a489f5d191cf36a4d480bab5ddc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
