)]}'
{"id":"openstack%2Frequirements~976789","triplet_id":"openstack%2Frequirements~master~I52cfe1e75ba57179d40e3167a45baf9f27ace956","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I52cfe1e75ba57179d40e3167a45baf9f27ace956","subject":"update constraint for XStatic-Angular-lrdragndrop to new release 1.0.2.7","status":"ABANDONED","created":"2026-02-13 17:57:19.000000000","updated":"2026-02-16 16:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5461b7b226f9c0226303e69fe426a42d48f8baf3","_number":976789,"virtual_id_number":976789,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:59:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b37ca82e460004b9fce39d9080578990f567eb89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69308be0e45d62222bd26cb428ca9cee4ba7b55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:59:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/820ffa3ad7744975b02cec59b57bdb73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfefaa3131a0494f8805567419227148 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d14c4de65414a55807c56fb60815644 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff8611b2618141c88b266d9313f1bd62 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d2f56d3c62e4535bc5853943827e0c9 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cc59412e7914179910b754f9ef859b0 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c0650c226854cdfb59a9f2d08c09997 : SUCCESS in 4m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9caeee74b8ff471e937643417cd6f181 : SUCCESS in 7m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0e29c94a3e7d4f028af7597d845113e2 : FAILURE in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9c81c3816e0b4664a18491f959f68edc : FAILURE in 2m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ea61bae2b2894f2297e06c01b5be09d9 : FAILURE in 4m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2d6b963e1d6246b8ba5781aa6df203ea : SUCCESS in 4m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/640ca2eff21448dd889365c71d6302d6 : SUCCESS in 5m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2dd55e8c188344b5b75d00f8aa2bd89e : SUCCESS in 7m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dbd6c1a3b45644ce9a36033f1547d180 : FAILURE in 5m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/92d8922f6b4c4ba2913eafce7f76e097 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/482c6579aff94180b84aa7afdefeb836 : SUCCESS in 3m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0992f0c959c84320b6518e66cc75dcfa : SUCCESS in 7m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3f9a64c7067740a7bfc74bb1c46aecbe : FAILURE in 2m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bebab09595e4160917cf4fc8d496cbf : FAILURE in 7m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a4b9f19497464ddeb9bbe673aac7cd49 : SUCCESS in 10m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e3403adcb5244ef48de8218a73d34bcc : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b5850b8dc58d4cd1a92279521d0f3fe4 : SUCCESS in 2m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/91f6658287814af0a9d3cb91adc2faa8 : SUCCESS in 12m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0cd9986ef1bb40f581b257d8bc2a5cdc : SUCCESS in 3m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2e691a9d4fb44b52acb31451cc15d2ee : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0e2aeceaf1da44cb9b1a336fa25b519e : SUCCESS in 11m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d6b9e4d4d22446039ffdd34f864dbc3b : FAILURE in 22m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/38f750f82e4d4d7888a47dc9625e16f7 : SUCCESS in 7m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8886fabc995e47a3863be2061a62bd31 : SUCCESS in 14m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/caa9ffb163d84e659451784a33b2c8da : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e77665152fa424299d29d211e552085 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78cd8f3249994ed5b53c59df261a6ca9 : SUCCESS in 28m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9bb9c52eba344cc7b90181fc28ecd872 : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/71d58f92629f45d4bda1d4434c5effac : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebdc3503591a451d8d52c94802bbd54a : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e83a52a14d2468092750d657505d04c : SUCCESS in 20m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e906fee3be9f421c9a3d325f83ea7b5d : SUCCESS in 6m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fe02736ee7c4079ab2d4d27e9a5be65 : SUCCESS in 1h 02m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c5ce30b0f574d9ab0505896206c3983 : SUCCESS in 1h 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5461b7b226f9c0226303e69fe426a42d48f8baf3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"175f02d410bdf52b380df94f7569149f6514b9e9","revisions":{"175f02d410bdf52b380df94f7569149f6514b9e9":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:57:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/976789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/976789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/976789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/976789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/976789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/976789/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:47:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:47:18.000000000","tz":0},"subject":"update constraint for XStatic-Angular-lrdragndrop to new release 1.0.2.7","message":"update constraint for XStatic-Angular-lrdragndrop to new release 1.0.2.7\n\nmeta: version: 1.0.2.7\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I52cfe1e75ba57179d40e3167a45baf9f27ace956\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/175f02d410bdf52b380df94f7569149f6514b9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/175f02d410bdf52b380df94f7569149f6514b9e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
