)]}'
{"id":"openstack%2Frequirements~976790","triplet_id":"openstack%2Frequirements~master~Ib84ea959a1e2361e93586c93feae1295b6ba8d7f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib84ea959a1e2361e93586c93feae1295b6ba8d7f","subject":"update constraint for XStatic-bootswatch to new release 3.3.7.1","status":"ABANDONED","created":"2026-02-13 17:57:38.000000000","updated":"2026-02-16 16:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d10174e35334095bfedf1a05be403d96b7e121f","_number":976790,"virtual_id_number":976790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:01:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf23dbdacc38f1cea213a9646e6849c0d076e3cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79cd902a014f153c7ef40f7705177028aa8ad6f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:01:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3942b0a75ff243da8f0ffb1ce8ca4558\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041c20b305684b8fb01335162b56cf12 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057b6450a0014e8a9e6e8d4d8922fa76 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c01e3794a06409083425e07df3728ba : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3202058b5e54c828ccbeee1009b6367 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bc4401656b247149ae4415de4792a71 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5cbe5397e6f64c6ab5e3ab95773fa0d6 : SUCCESS in 3m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ec49a559e306403e958dda44886181ad : SUCCESS in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2f6a7e7b36c744b389a3fdb51d5fa209 : FAILURE in 3m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/59e6e821967f4c1eb5a90065fdf88328 : FAILURE in 6m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1be6c1dc918743e79b156eddc51aecea : FAILURE in 4m 57s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fdce37adff4647a28e794bc6393265b8 : SUCCESS in 3m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a99195bc25f741c3a73de0f0f60e0ac9 : SUCCESS in 5m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e9e64f7e2a854aff98c97c6d0a943489 : SUCCESS in 6m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e1cabb864de40d9a1d7e850004e5c2b : FAILURE in 5m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/684e8955e2f048bfbbd38e5268ee4078 : SUCCESS in 6m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b2a0a267fa1d459d80b3c2e1928d16ad : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5a0644c45690458b80cf394cd217163d : SUCCESS in 7m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/05550469fee34f5c97511eb871ede251 : FAILURE in 2m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8dfbf7a8e694abdbdb6a9926a42f47d : FAILURE in 5m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0ebb9497a6cb4196b5617cc24a2eb33d : SUCCESS in 6m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36b3f3a517e341649f2df23b1a7c4e09 : SUCCESS in 12m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fff25061f3dd47b0b401052d2b0513b8 : SUCCESS in 3m 38s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/91b061344d7b4cc0b799913e4620faa0 : SUCCESS in 12m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/44da7da29fc347a383f8e60db2500b6f : SUCCESS in 6m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/db5c16b10799422d931a99c473ae25da : SUCCESS in 3m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e29346f7184a4595bda1e03f366228d4 : SUCCESS in 11m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/051e813153614ff999fcc285a4b4f158 : SUCCESS in 26m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/28b5a170f41a480baa1c747a36178b1d : SUCCESS in 5m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/74f215211f8d48daa5794168c02995bd : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea1555d0656c4ef78e6937f2a808e7f3 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f33a11cda520448eb1e358a074ca3742 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03abbff392e44fbcbc7f0f3db2161843 : SUCCESS in 25m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5ba7580ac64049849cd3e46cf8fc251a : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6828e1886051493eae2359370215f3e8 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b050f3729b7b451d802594156ecc9d17 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ed1c6b456a504dd68df9f923584ac331 : SUCCESS in 17m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a03bd52d7f0b47d59e9152713f503487 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b1fbaabd01f422a8189b5bd1f2e4a3d : SUCCESS in 37m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/763b587e6f2f4de7aa89ef02b721f8d0 : SUCCESS in 1h 49m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3d10174e35334095bfedf1a05be403d96b7e121f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"382cc0d0f5bf9a241079c54c162e0040b3846072","revisions":{"382cc0d0f5bf9a241079c54c162e0040b3846072":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:57:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/976790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/976790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/976790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/976790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/976790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/976790/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:47:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:47:35.000000000","tz":0},"subject":"update constraint for XStatic-bootswatch to new release 3.3.7.1","message":"update constraint for XStatic-bootswatch to new release 3.3.7.1\n\nmeta: version: 3.3.7.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib84ea959a1e2361e93586c93feae1295b6ba8d7f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/382cc0d0f5bf9a241079c54c162e0040b3846072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/382cc0d0f5bf9a241079c54c162e0040b3846072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
