)]}'
{"id":"openstack%2Frequirements~976793","triplet_id":"openstack%2Frequirements~master~I12982c5295261972fe73f2f2aa38fbc412c55447","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I12982c5295261972fe73f2f2aa38fbc412c55447","subject":"update constraint for XStatic-Angular-Vis to new release 4.16.0.1","status":"ABANDONED","created":"2026-02-13 17:59:04.000000000","updated":"2026-02-16 16:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"653f17b54967b3231367211f70a76584e6df2585","_number":976793,"virtual_id_number":976793,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:58:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfeb68c527a8c519bf426e6313b45b194dfdd580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96297407d1c412fa5067515eed5cef0d723d2d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:58:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cc0a8973e7741d69afc1bd0dbcd88e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977b6171141347cd883f54208557da41 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b46fa1c899f45c3899465acc5771c85 : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a48a1f2e56849e48407bea8ed975d36 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/361522d32e9b48008cd4c8b065bea9f5 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/634781c0c15e4128a45438e11a67af3e : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0efad6b398eb48ce9f4a06ebb3b76c57 : SUCCESS in 3m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9658e246244848589b2c59fdd3ab47c9 : SUCCESS in 5m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cef5715844ff4ced8b7debd1c7e2a0c7 : FAILURE in 4m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0bfd7e0ded80427399a4719167efbedf : FAILURE in 7m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ddd373a6e5af4d6a9fb60c9f30203203 : FAILURE in 5m 40s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b137451406d0412d930fe250d915e7fd : SUCCESS in 6m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/25dfa1a941644adbae969d9bc9befab6 : SUCCESS in 3m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/499a42b9fdef4ba4b8b70e0b122f4ad2 : SUCCESS in 5m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/725397a1d0a74f39993971aad278b8ac : FAILURE in 3m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b766ec0bb939466aba4759df55962216 : SUCCESS in 5m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/17f4353ec19f4107859959b675e6726e : SUCCESS in 6m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/53da41dbee014b8f93fcf3d5782448d9 : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/010132010dc54d61b0fa2ca48b559d95 : FAILURE in 2m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e12a8baef6748adb08199c9e1876ae9 : FAILURE in 6m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f17ef4389c1f46a4a61464e447e84c6e : SUCCESS in 8m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/db22d948b66d417ba083980e011785d2 : SUCCESS in 7m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b3153ece8d24ab58a1314fd632ece76 : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bab8ca288f834cdbadcd0dd04296ada1 : SUCCESS in 9m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7e7f4b5c26ca4b818a1ea3ab1f224b68 : SUCCESS in 5m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/81b2f1d599a44fe5a3eee3079b527a37 : SUCCESS in 5m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/28e6f4ed55f54ba084825ddf9cba2b39 : SUCCESS in 11m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e04b21b4d294d0fa1194f6e5a222218 : SUCCESS in 43m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c677f25a8aec44789ca006d21b2259ef : SUCCESS in 7m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1787165ddbe7474684e5db82cc05d491 : SUCCESS in 14m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9170d0527a6463aac2985d0c2b29957 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36a7cc09e00944289e7b7bb7e8a864be : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7db4ad084789402d94c658bc6b3ba969 : SUCCESS in 27m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/160f13f4c42b4ac38cd7c58a12b4bebe : SUCCESS in 7m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dddb6ae3be0c4b348e7a3bc95663ed28 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37ae4c42f83746f8bc5c5be6d3bbee3f : SUCCESS in 5m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/be43bded0c6740b2a9a015fe92eff3d4 : SUCCESS in 19m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0ef85a14a3944488a999073aba64df1c : SUCCESS in 6m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/512e890d4fd24888b4957cb3c4d09cd4 : SUCCESS in 1h 07m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de81167f1bad401b8ff9c44ed65e53ff : SUCCESS in 1h 46m 05s","accounts_in_message":[],"_revision_number":1},{"id":"653f17b54967b3231367211f70a76584e6df2585","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfb174edb0f2adf359f846ad701df4e3034cb3cd","revisions":{"cfb174edb0f2adf359f846ad701df4e3034cb3cd":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:59:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/976793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/976793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/976793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/976793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/976793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/976793/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:03.000000000","tz":0},"subject":"update constraint for XStatic-Angular-Vis to new release 4.16.0.1","message":"update constraint for XStatic-Angular-Vis to new release 4.16.0.1\n\nmeta: version: 4.16.0.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I12982c5295261972fe73f2f2aa38fbc412c55447\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfb174edb0f2adf359f846ad701df4e3034cb3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfb174edb0f2adf359f846ad701df4e3034cb3cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
