)]}'
{"id":"openstack%2Frequirements~976794","triplet_id":"openstack%2Frequirements~master~I89d55d3d1da8989b645b7c02103a1b25630a8d14","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89d55d3d1da8989b645b7c02103a1b25630a8d14","subject":"update constraint for XStatic-Angular-UUID to new release 0.0.4.1","status":"ABANDONED","created":"2026-02-13 17:59:34.000000000","updated":"2026-02-16 16:22:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8bf233152b1fac6248267717dd73a417784f9f6d","_number":976794,"virtual_id_number":976794,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:25:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"896757639a2ffb4390775e775dcc9ff2af05a721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22cf09a68a39f56bd66fe389be043ca675701b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:25:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3588ca08adb04f9f88c37d2fed122656\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eee84e3411245709739aec270b93372 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01f6762cd51409e9d6cc6e6740391c3 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50eddc96a61740deb11e3a757713b7e3 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b82848564144e8ab21e4b80ce2d17e1 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0354ead64e10423a9bd68e575b70ff9e : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/421589b3ecf6494f8bb7010588d4e57f : SUCCESS in 4m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/90cf168845c44be2bf8927cdd5e00cc0 : SUCCESS in 5m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7ed2dde82631479dbeb4bd3b3d3ca9a9 : FAILURE in 7m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/98f1bdc880fa49f79cf8a73506501632 : FAILURE in 7m 00s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/73984967f89140839d9b74cc0f61d4e4 : FAILURE in 4m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ebc93206000846539d4a00888c241681 : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c193d26a9e444581806dcc7b5f05dddc : SUCCESS in 6m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6a554b9886254f9e89ac3c3d1cbe155a : SUCCESS in 5m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0a7e3ec2ca024984a2b743f895bd2af7 : FAILURE in 3m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1bdb014b49eb454aab42bb6c4962a485 : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/26ca444b586f4547895f311c60560337 : SUCCESS in 6m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/78933b857cc7443cbe6e60969d59c3bc : SUCCESS in 9m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5964e4a3d4a94236a9796f635af68a97 : FAILURE in 3m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca395065bbe842a9a140268a5085c25d : FAILURE in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b5e76fa8bd894a24865ac973a2a77c63 : SUCCESS in 8m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c2bddfa077834bb99b91395e9d3751b9 : SUCCESS in 12m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/443d2d1f77dc43afa25b755b100c1d4e : SUCCESS in 3m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9de41778b35e4ca98597528252bebfa2 : SUCCESS in 10m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/17ca61da4b35419cb6018ab33ccbd8ca : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1a1b76ec15c54273b65d649d7a6c1a72 : SUCCESS in 5m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/00259fc4a5cf48f69ca5b55171533464 : SUCCESS in 10m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/587d4fde219f43459e08fb14fbb8c146 : SUCCESS in 26m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bfd119a6bb6f4462bee6283a36c74822 : SUCCESS in 9m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/87912c20806e491e972e2d0100cf0036 : SUCCESS in 14m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7bff7f7857c34ab79a8285651b5b5f79 : SUCCESS in 3m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f5549e6ed8c446fb032c0c817a61f34 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f9d344e1cc444bdb12a043791308559 : SUCCESS in 25m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4776a1df01344a309bb6517c7228181d : SUCCESS in 7m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2a02101e20c343d590d8b9cbefc10f7d : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09556d5160bf4578ab468db8a52320b6 : SUCCESS in 4m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bb67cc2738f145b6a6a8e334dd67f8c9 : SUCCESS in 19m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/649be43b1e2a4749acdc16faf32a2e83 : SUCCESS in 7m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8712de255e2469eac03dbeef55d45a1 : SUCCESS in 1h 01m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/052967de6bb942c8b8966d7b9e8d36ee : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8bf233152b1fac6248267717dd73a417784f9f6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"831b2191e398db68f0c58cde3c50bc114efc20cd","revisions":{"831b2191e398db68f0c58cde3c50bc114efc20cd":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:59:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/976794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/976794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/976794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/976794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/976794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/976794/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:33.000000000","tz":0},"subject":"update constraint for XStatic-Angular-UUID to new release 0.0.4.1","message":"update constraint for XStatic-Angular-UUID to new release 0.0.4.1\n\nmeta: version: 0.0.4.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I89d55d3d1da8989b645b7c02103a1b25630a8d14\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831b2191e398db68f0c58cde3c50bc114efc20cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831b2191e398db68f0c58cde3c50bc114efc20cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
