)]}'
{"id":"openstack%2Frequirements~976795","triplet_id":"openstack%2Frequirements~master~Iad289836b588406296b9c70d7ccdc177cd99cde3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iad289836b588406296b9c70d7ccdc177cd99cde3","subject":"update constraint for XStatic-JQuery.TableSorter to new release 2.14.5.3","status":"ABANDONED","created":"2026-02-13 17:59:53.000000000","updated":"2026-02-16 16:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5413e91e320c45298b0b38621341197fefa943da","_number":976795,"virtual_id_number":976795,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:09:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:09:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adcacba610412ce7cd96c35083450e17451871eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 17:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eae773cdb7dc4cad28ca4d4290aad8c6ab3d6d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:09:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1d6fdd95b8d4661a5dfddce0d5bf941\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6998ca311cf746279c3ed38b167a28f9 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c6be913f354537b8106e6ed59beea4 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a19fbb6a18a84346b0725ca616a21f11 : SUCCESS in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b89f82f563d4de090f2f1843f593b6b : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c587b084bb494a71aabea2d5e3db011e : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6464971e5d604c09b815f7b61ff8a1d3 : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a1ce16d93fe4ffb9c38ef34b27c486d : SUCCESS in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0972d5c6f3e14388bc146b5a852fb965 : FAILURE in 7m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7886bff4848547079dde518804c28f51 : FAILURE in 5m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/da4bad1936ad400888e91064a6d7471d : FAILURE in 4m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4abf0e13ed424dea96144b4a28318bcd : SUCCESS in 6m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9229fb2263704ea8b8653e53d3d13a06 : SUCCESS in 6m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5be6fbe5260d40a7a7369b8f62386582 : SUCCESS in 3m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/24bdb4c12e85459b879403ac3fb06648 : FAILURE in 6m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6cab641e54ac497db62cafd7f32bfbf8 : SUCCESS in 7m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ee975b63a5404bfa94c71c3984cc2f28 : SUCCESS in 6m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ef9301b65c3f4851b8e3c83f73a2da50 : SUCCESS in 7m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2b2423b5e7284168814342a2341a93f3 : FAILURE in 3m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e648de3eae148ffa731790ecaaa3b81 : FAILURE in 3m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2ea7cbb6a4dd4ecfa2561e79ce924710 : SUCCESS in 6m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/85641923ffd6458cbd89cacec5288fe6 : SUCCESS in 12m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/18c7a6a716214acabab8dcdd138b99ff : SUCCESS in 2m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8756f7685b6b4acc8311de82d793171e : SUCCESS in 11m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e897f86bf91a49e0add896d819bc8cc0 : SUCCESS in 3m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9cc79295bc464c5c8f5e7b0c34402fc0 : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4d89dd335c9045dca31bfafc06247357 : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8d906f4bfe97446e828ab34704f69e58 : SUCCESS in 23m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/30e201ce88504585934a160304798f93 : SUCCESS in 8m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/869b681159ba44b1b5dafcc55c69016f : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c0154149f517443fb7e0524e71b54dc2 : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6f0dc9118924f3293c04eca1489c268 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1eebcaffa354f7e8a5c2c931d127f29 : SUCCESS in 31m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c55f1970cb4e467aa25545ec59f52599 : SUCCESS in 8m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8c9dade870a1441eaed73ecc72053582 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a663c1405f1423287ee0d52206062c1 : SUCCESS in 7m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3e8afcc1f6b431aac5411bc881f9b78 : SUCCESS in 18m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ccd97fddacff408a88a86fee7938ad31 : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abdb33256eba41c197034e6076777691 : SUCCESS in 1h 08m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb2788b57bf64bbaa4521cc505db16ce : SUCCESS in 1h 55m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5413e91e320c45298b0b38621341197fefa943da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f841695c8df459918955a8c0fbbab5f122cc4e39","revisions":{"f841695c8df459918955a8c0fbbab5f122cc4e39":{"kind":"REWORK","_number":1,"created":"2026-02-13 17:59:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/976795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/976795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/976795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/976795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/976795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/976795/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:49:51.000000000","tz":0},"subject":"update constraint for XStatic-JQuery.TableSorter to new release 2.14.5.3","message":"update constraint for XStatic-JQuery.TableSorter to new release 2.14.5.3\n\nmeta: version: 2.14.5.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iad289836b588406296b9c70d7ccdc177cd99cde3\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f841695c8df459918955a8c0fbbab5f122cc4e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f841695c8df459918955a8c0fbbab5f122cc4e39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
