)]}'
{"id":"openstack%2Frequirements~976796","triplet_id":"openstack%2Frequirements~master~I4c1ba6047ee3788667071557b3ed1f3373cea456","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4c1ba6047ee3788667071557b3ed1f3373cea456","subject":"update constraint for XStatic-JS-Yaml to new release 3.13.1.2","status":"ABANDONED","created":"2026-02-13 18:00:17.000000000","updated":"2026-02-16 16:22:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7857657a15d07efed8dbb49ebd8bf0c6676b9a0","_number":976796,"virtual_id_number":976796,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:54:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c77cbcf070e015f4cb8c57c0b48ade41ebf29757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 18:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee73e574aafa0c7d28914278485ca007deaf1067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:54:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c61007a6da7a436888e06bd236c446b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68430829459f487083f7d04d9a0662e6 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7709d78bb51c440f88214cb661d5075c : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f6d52d17a3f4ced8463925f9b38d466 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/320eb270b0e24bb89d942fa4790ee2b3 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fdfdd34d94a43e7baf9a21ce94b055e : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1586909359c8472d8811671cc61c4714 : SUCCESS in 3m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6115fc4dc6254b4e92c4074be7bc3cd5 : SUCCESS in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4b82596526484325a9d35daecd370034 : FAILURE in 6m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0e703dc8ded24b79b3bcc60626da1d3c : FAILURE in 2m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/92d75fd5811f4ba494670eba39d32dce : FAILURE in 2m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/151ba47250a14252968f7eeb6da05af0 : SUCCESS in 3m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cfcf72910b134a00a3f985e3a30fb4ee : SUCCESS in 4m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ec3ef7dbbf42464e913bd62ce75055e9 : SUCCESS in 3m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/75632a90cc074a258610704fa4788509 : FAILURE in 5m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/71fa561013b64897806981e494ac4711 : SUCCESS in 3m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a60ed7f5b3be442a9ed357d923af2356 : SUCCESS in 5m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/67412354e7db49859e9625801c1835f4 : SUCCESS in 9m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f5008ccb9a7d45fe8f831603741413ae : FAILURE in 2m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16f5890c9b2740d9a25246d49992f87e : FAILURE in 4m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a823120241214ba6b2bcf4648faace51 : SUCCESS in 5m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f595f7d8d0a84194ad6baac014cf509c : SUCCESS in 14m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3c6c0c95858c4dc9929a765caaf03e55 : SUCCESS in 3m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/76b4ac39af134f31be5a73e0ff6e26f0 : SUCCESS in 12m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/54f8fc7bd857472488ed37c50bcfd5f7 : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5abbaa1940ee4dc68d550e8175f05261 : SUCCESS in 6m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2007f50f93664f5f8b5cbdd21ad1f3a5 : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c11627d0a50c4ccea369e559251bf9fe : SUCCESS in 30m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d526beeba516430b9c2c21e3e2fe08c0 : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a25e1c91c0864ddc90eea9291b7e6726 : SUCCESS in 13m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b99887ca46f64738a1cb574b72c58e37 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9b4cffa7e06402493eef449c44c179a : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad58e13ea9e14e4c9994664b6975d1c1 : SUCCESS in 26m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f9a0f016faa64df4a4d1b3028634c1cc : SUCCESS in 7m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8a118df01341443587e55af077dcbcb0 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90b455c05bd84949ba35169e323f2204 : SUCCESS in 4m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d857ac8908204907a23f5fc09b0fa518 : SUCCESS in 18m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9343706988c3401dac2819e041145341 : SUCCESS in 7m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed469daff5094d08a0e19ca8766ceaba : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efe38247e69946f78dc277593c6bad3d : SUCCESS in 1h 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f7857657a15d07efed8dbb49ebd8bf0c6676b9a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b0b28bc25495827fa1561f0e862f9cd9e87a97c","revisions":{"2b0b28bc25495827fa1561f0e862f9cd9e87a97c":{"kind":"REWORK","_number":1,"created":"2026-02-13 18:00:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/976796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/976796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/976796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/976796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/976796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/976796/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:16.000000000","tz":0},"subject":"update constraint for XStatic-JS-Yaml to new release 3.13.1.2","message":"update constraint for XStatic-JS-Yaml to new release 3.13.1.2\n\nmeta: version: 3.13.1.2\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4c1ba6047ee3788667071557b3ed1f3373cea456\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b0b28bc25495827fa1561f0e862f9cd9e87a97c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b0b28bc25495827fa1561f0e862f9cd9e87a97c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
