)]}'
{"id":"openstack%2Frequirements~976797","triplet_id":"openstack%2Frequirements~master~I835bde63c58b88c5a6aaa5c194372a5b97cbdc66","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I835bde63c58b88c5a6aaa5c194372a5b97cbdc66","subject":"update constraint for XStatic-JQuery-Migrate to new release 3.3.2.2","status":"ABANDONED","created":"2026-02-13 18:00:28.000000000","updated":"2026-02-16 16:22:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd62ac8b5fdad72ec666b949df64b4e725b7bfd7","_number":976797,"virtual_id_number":976797,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:19:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9c567bb8942f18dd3d5d18e2b2b34e3b7d4e132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 18:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2badb3ee906c477ab3247d3d09947644fcbcd781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:19:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/466f1009e0d64a639d3ddf0f37814576\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2bc7c52656240c68dae3c34ae2b977f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d7097dd9aa477e85a1aa212804b556 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/150b4dda22f6442590b2d816d0b9a886 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c237589335ef40ba8d9d427d7436545d : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85f4b81afdcf440ba4ebd26801c2a5c8 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3eb1f369e9d34506ae8994fdd14bb777 : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3767d8ab33354dc49f21c9b5ecb7620e : SUCCESS in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f6bd1a16c23c4f4d977ea88912398cfe : FAILURE in 5m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c35166330c2b482680064934dcf0930d : FAILURE in 4m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ce9d08e5d7cc485aa43e97438b55288b : FAILURE in 3m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fe1201cb172d4101af467b6955fb9a5a : SUCCESS in 7m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8a204c7f28394a8b95e68a1a3c8405c0 : SUCCESS in 6m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/48a465cfba2f444487ffcd0bb35f1b40 : SUCCESS in 4m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8ddf50af1ab8424c9c6ce6a7f2e0861d : FAILURE in 5m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d23a3ba37bfd48759d117ac894a7dbf6 : SUCCESS in 4m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/72cfdc1288244dc28562cd331dfd15b5 : SUCCESS in 6m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/23c3988f14d3470ba251ce6b10d37e9f : SUCCESS in 5m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cf7bc6b3ca284e5f99d33d30cef1ac7c : FAILURE in 3m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26e57b0278564c20b157730bd68a406b : FAILURE in 5m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/492df8ff710d45be9136c6ba42147a9d : SUCCESS in 9m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/562cd80777d04581b8e979dbe38fc354 : SUCCESS in 12m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7c268e799b0346248028800968ac0bd3 : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ee9541d5669c4c3492f52deefb3c726b : SUCCESS in 12m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f54cb04c511e4f42bd02cb60b1db0fcd : SUCCESS in 5m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7f0b0ea78f5e46d2b98ac39633d051a3 : SUCCESS in 4m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dd497a50f6764c2babe3a4ad130ed83a : SUCCESS in 11m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f819fecc05ba4cc397b6f0d4dbab0c0a : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/de550f693a694bbb9f9b703383f930ee : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/42c2c6d79bae4469aaa82ef942104d20 : SUCCESS in 15m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ede34a7cff7e49f0bbacfca374f5e59c : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af0429a697364923a5d66c46603ebc5a : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d817b031433440cda04730dd2b861bdd : SUCCESS in 26m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1e44ae3c713e439db64c432a5914b9bd : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b7bba36841e5409e873142888487f7dc : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fcf7f65d52143bfa194b0194ec66c40 : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9da06f8ba65e43698bf1071ea0d8b0f0 : SUCCESS in 7m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/65aa9ec646854b99b8e72a92c019932d : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa2b7ca431404a29b244dac4a108fe8c : SUCCESS in 1h 04m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcd0aa36567d40119b81d33d0788bc05 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"fd62ac8b5fdad72ec666b949df64b4e725b7bfd7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f41c2fb4f3ace4dd2db6300fd25de541ccdd34a","revisions":{"7f41c2fb4f3ace4dd2db6300fd25de541ccdd34a":{"kind":"REWORK","_number":1,"created":"2026-02-13 18:00:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/976797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/976797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/976797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/976797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/976797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/976797/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:27.000000000","tz":0},"subject":"update constraint for XStatic-JQuery-Migrate to new release 3.3.2.2","message":"update constraint for XStatic-JQuery-Migrate to new release 3.3.2.2\n\nmeta: version: 3.3.2.2\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I835bde63c58b88c5a6aaa5c194372a5b97cbdc66\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f41c2fb4f3ace4dd2db6300fd25de541ccdd34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f41c2fb4f3ace4dd2db6300fd25de541ccdd34a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
