)]}'
{"id":"openstack%2Frequirements~976798","triplet_id":"openstack%2Frequirements~master~I467699e242bee1fb0d35f609824a9c153360aae9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I467699e242bee1fb0d35f609824a9c153360aae9","subject":"update constraint for XStatic-mdi to new release 1.6.50.3","status":"ABANDONED","created":"2026-02-13 18:00:58.000000000","updated":"2026-02-16 16:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9406c21a00234e2c6f0d4db93cd814e11043e58d","_number":976798,"virtual_id_number":976798,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 20:00:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 20:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c6312df957a55b4f5e211ed9fd92549683c49ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 18:00:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"758f53aff4deab42a6832dcfd4e5dfb8dd00fe0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:00:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18b4c50f6d684473b7812badb48ba7d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6438df434b14bfa9f07c4c2163c3148 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978dd719da19420086fd41e9ad4f14aa : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c41d24975b04f61b024c058320235c6 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/877b9c518df346c49f8e78b486db1034 : SUCCESS in 2m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed752dc1512f47c0bcbae189579572cb : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d9b9008239140428d6e6e0c5fe258a7 : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/be8940202c984ecbab5cf185f0bd937e : SUCCESS in 3m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3a05b66f34643eaa905a6152efbedc1 : FAILURE in 5m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/00c9d047f8ef4a64b34cb3e9b018c114 : FAILURE in 4m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3205c2a4df2d4b039e4032b7d140c20d : FAILURE in 4m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c2e054df79424d5c90faf13e48c3a4fe : SUCCESS in 4m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b89ccd739fc147c2bd36b53adbb4b74c : SUCCESS in 4m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5b86e3ba8bd740459464aa289fb87f98 : SUCCESS in 4m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/97a005534be5481d9e1de7724678fbff : FAILURE in 6m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c4924fc48954429fa285ce992d2a71c4 : SUCCESS in 6m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/21bd4ca25a65414b99e0b7c48201c63b : SUCCESS in 6m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/569eb61056dd41bc926ecb8530a6e09b : SUCCESS in 7m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5c7d4539c2bf445bb7372769a2bb0917 : FAILURE in 3m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3065de7cc33d48cf817386789fcfe900 : FAILURE in 3m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c3e67602de5e40b8acc4da9aa1674fc0 : SUCCESS in 9m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/74628afe08b246c38619c8cd4f403e65 : SUCCESS in 12m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/56dba98e36cf4ff7b02517599c1bf063 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4ce8fb0b3ac6413f96ce400d6edc558e : SUCCESS in 7m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f8ff1d5ddea3496d8c205be97333dc30 : SUCCESS in 6m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4c43ce75270d426981cfa8d0facc12ce : SUCCESS in 5m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6cac66a3ae324078b8be0918c7cc5d64 : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c480cc0e3d6848d4a0f7026a2387381a : SUCCESS in 18m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/97889ac9669f457d8f9903d34d2cd699 : SUCCESS in 9m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a85956856b5f4b9f8ff3c7a8ebe346dd : SUCCESS in 15m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/90d6ab0920844381b055de2e75b8c1b0 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a947afbc46214330bfaff03bd1017954 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23c1d6c265824cf893c035a779b77640 : SUCCESS in 25m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cb7f02250ce541b8b443c6f9d10ed38e : SUCCESS in 6m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c6663b78bfb54307a4f4587f50bc41d8 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2914f36ebeb403f9297d7210d456775 : SUCCESS in 7m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b14c98481c174dd6886ee4f3f6d0de48 : SUCCESS in 14m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8d236fbf85de4f54a95cf054819fe0b3 : SUCCESS in 7m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e30b514e822244aab306b719d221daa1 : SUCCESS in 51m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6137c9dff1a94b6192e929798fd16f1b : SUCCESS in 1h 47m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9406c21a00234e2c6f0d4db93cd814e11043e58d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e91ff2e9882429a9407f3b8e3c6523a2a6fdc6fa","revisions":{"e91ff2e9882429a9407f3b8e3c6523a2a6fdc6fa":{"kind":"REWORK","_number":1,"created":"2026-02-13 18:00:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/976798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/976798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/976798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/976798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/976798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/976798/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:50:57.000000000","tz":0},"subject":"update constraint for XStatic-mdi to new release 1.6.50.3","message":"update constraint for XStatic-mdi to new release 1.6.50.3\n\nmeta: version: 1.6.50.3\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I467699e242bee1fb0d35f609824a9c153360aae9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e91ff2e9882429a9407f3b8e3c6523a2a6fdc6fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e91ff2e9882429a9407f3b8e3c6523a2a6fdc6fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
