)]}'
{"id":"openstack%2Frequirements~976800","triplet_id":"openstack%2Frequirements~master~I5c63e5b1c87b7e969b8725e08e29e2a01dd572d2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5c63e5b1c87b7e969b8725e08e29e2a01dd572d2","subject":"update constraint for XStatic-D3 to new release 3.5.17.1","status":"ABANDONED","created":"2026-02-13 18:01:38.000000000","updated":"2026-02-16 16:22:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4760d3ff8c4731eb25ae7410151223568c198a20","_number":976800,"virtual_id_number":976800,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-13 19:45:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 19:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d852a57fc68a526dc970bfbccf7b0e1840574d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 18:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9886562d1d1d68b91d4360b1ea3092f839aba18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:45:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a291dcfd7840beb2826ef23a98c1f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0f005a55cf4ccfbcf1c2edba3d379f : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18bd101a3a614db29e8b34bb6f90a60b : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a88300b31064794a961e6c5f7710f97 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbd4f4643262407cadbd2db9f0d24200 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93ba7afb19ca45b8b3056c239a3db97b : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/094c2ecb724549adbc15a28040c446ed : SUCCESS in 4m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/467189b739b84a21aa3e6cabc52a4404 : SUCCESS in 3m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/21cc00bd78c049209d757c101b0589a8 : FAILURE in 4m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fe5d069ba84d42a4b343331f1c6550c1 : FAILURE in 3m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/eec6d1f1a6584276bdffcf901500ccc0 : FAILURE in 4m 29s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d86527d95bdf463eaffc932b7c95fb27 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6a6b4b9c4f3c457bbdd4e1ba5c2d0830 : SUCCESS in 7m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0884770cae054e22b4da4a0a468f28a7 : SUCCESS in 6m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4c8fef417d7b4ef5b4081c8146cd93c3 : FAILURE in 5m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f974b2361de14f2abe2b01cae8f97c7f : SUCCESS in 6m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4c3ff004d46341b9a7ee436be9793748 : SUCCESS in 3m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2d4aacc47fb44544bdfdc49fb619524d : SUCCESS in 8m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cc98aea5647a4f9ab8d982b864f651c4 : FAILURE in 3m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ad33ba111664995be9cd608439b0e7d : FAILURE in 3m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/10a9076762cd4a568a27d613e42698da : SUCCESS in 10m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/890da75dc82a4c938d588c5edcd63e5c : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/da26f7b946ab418689898ccd3ec33dc8 : SUCCESS in 1m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5fa64361adf84321b82556faabd20155 : SUCCESS in 12m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ba92a885caac47658a7a78d7d591f521 : SUCCESS in 4m 35s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/58455ce9268741198678761012da8056 : SUCCESS in 3m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8c6d221203154bb4a238e5884cb7269b : SUCCESS in 10m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/60168b859f82461c91dd0b452ae53c4c : FAILURE in 23m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/990879a174264917aae77f7a41709282 : SUCCESS in 4m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/dcd2b894d77143ae80348fe472e4ab5b : SUCCESS in 14m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e2415096d1ca4e71a5a2c14a6f33d4ec : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c5b4cf2dc1145fcb7c99ad96c0afb58 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c1e6ec36169401f8522643144a94782 : SUCCESS in 25m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cb4afc1472874d4b92b0726e40d0030e : SUCCESS in 6m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/669142b6eedf418bbbf2d983b088b0cb : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7e166fcdd0746c595387bd5dcbaecdd : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/92d35853234c43a38d5d4617986bfcad : SUCCESS in 19m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6e02baef51a44b7588b387ebfe005f6a : SUCCESS in 5m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/531da25db9ce4d9c883ab7b21f8c0c7a : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d03baf3c8658448d825bf9dca70d19fe : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4760d3ff8c4731eb25ae7410151223568c198a20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6578e8a51fc82c2914432e120adf4969274319e","revisions":{"c6578e8a51fc82c2914432e120adf4969274319e":{"kind":"REWORK","_number":1,"created":"2026-02-13 18:01:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/976800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/976800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/976800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/976800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/976800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/976800/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:51:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 17:51:34.000000000","tz":0},"subject":"update constraint for XStatic-D3 to new release 3.5.17.1","message":"update constraint for XStatic-D3 to new release 3.5.17.1\n\nmeta: version: 3.5.17.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ie54be0b4259db7e7fabf846ad72ee7d03536ea06\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5c63e5b1c87b7e969b8725e08e29e2a01dd572d2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6578e8a51fc82c2914432e120adf4969274319e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6578e8a51fc82c2914432e120adf4969274319e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
