)]}'
{"id":"openstack%2Frequirements~976811","triplet_id":"openstack%2Frequirements~master~Ifeae726aad5171192c45aecafe198715b416c60c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifeae726aad5171192c45aecafe198715b416c60c","subject":"update constraint for tooz to new release 8.0.0","status":"MERGED","created":"2026-02-13 19:55:59.000000000","updated":"2026-02-16 22:49:35.000000000","submitted":"2026-02-16 22:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976811-new-release","meta_rev_id":"62556db1eee92c78f12e5095d7e3d67a7189afbd","_number":976811,"virtual_id_number":976811,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 22:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-16 20:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-16 20:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 21:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 17:14:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-16 20:56:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1104dff497d143211029d7d0a7527eb91f4b645d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-13 19:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"142ced4eda5ddec5abcad762bd55a73d507e0255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 21:23:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0690a560583241d9865dd45ee6cc1155\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8b83868eaf4b4989210f42905290c8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624c32a593a04555886021e933cf13d6 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4e28248ad0941a9b58b6cda1ce02c4c : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcdb45df15204ec697763134484e801d : SUCCESS in 2m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/add2d0feca8f47bcbf83378df47e229c : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/597b9c82943c4b909369d35ff9667743 : SUCCESS in 2m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f151d3d695c242fc89f944b054d41589 : SUCCESS in 4m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/24bbdc3a7a864f49a553e85803c8cc3d : FAILURE in 3m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e0fcedc44c6746e9aa4a9c31af066646 : FAILURE in 6m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4a2d15853862446e9d6827c8171d7afa : FAILURE in 2m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b78684327362468389a4683e6b55b1a5 : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/075c66fcd2d44f17bdafb1a97bf94031 : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/18e77fff77c7477a8866ee03d5aa389d : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a90accce5c934b7785b1ed4823db6c79 : FAILURE in 3m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1d1c7a9e2fa74ceeaa352adc05add6e5 : SUCCESS in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a56c16ec7f034f28a21e8578a0a0c139 : SUCCESS in 4m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4f4f355dfdc34d8cbdb669dd9890d311 : SUCCESS in 8m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1913c284c515406881222005a9790aea : FAILURE in 2m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/520c65fdc5d74ac5a7ba3b881a177476 : FAILURE in 2m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/236147d105c14422b5f54ae7d6b0ab9c : FAILURE in 8m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c5297c06c4c04cca97ea209bea0cf9f6 : SUCCESS in 9m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5716c45fdff14f52b3f3321bcf146369 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/04a6323cbe9c4307b527470c3d0315e6 : SUCCESS in 10m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/69605b081abf4de6a055eec117888eb4 : SUCCESS in 4m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dacef570d5434e2f9a8ac6599be7b18a : SUCCESS in 5m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7139283eb1ac448d8a5d6fd704964f5f : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6876ba78e155403cb97bcf2a27d162e9 : SUCCESS in 18m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/54b6e828b5cd4bda864269e1ad779798 : SUCCESS in 7m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d10f497dc85a4dafba617d2193751932 : SUCCESS in 13m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/30a9de18ec194e7fb2441a3687bf0a35 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/333c6937e66d450fb98f21d8d0ca2d72 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f51157668c7f492195b47b169ff06def : SUCCESS in 29m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/412ca802814a46ed861176b6465bed49 : SUCCESS in 6m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9cf57fc9447f4326be8eb54c0d736ece : SUCCESS in 3m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/393919ea944b4127849e03356fcaaa32 : SUCCESS in 3m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/33828ecffac9402986e1231cfe6e6c2c : SUCCESS in 20m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5ce3434a68db45b48671674230c38c9b : SUCCESS in 6m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c5d11f9dd37491bb84816c169fc89ce : SUCCESS in 32m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30ca33631c544fa5ab6d3237d536c5ae : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":1},{"id":"88b23744977535028b6067afeef539250e6217a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 17:14:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc0a79a7d7a9b6af5fa8a24e23d5031d1cf24d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:15:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a5df4a2c7b4d8eaa5af8d6b87e09a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29745b02312143e6ab58e296239b5fe9 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee49d239e444991b4131f4c171fbe30 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/585ead77f3334d83b04a405f7d309658 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e2d37c58a1747a59c880ab6076d6578 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b7b681779414ddda603e412da048695 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b07acacaee540c2863ccadf902b9ae6 : SUCCESS in 3m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/86e1602a4ae6453186477f9db1c9c672 : SUCCESS in 5m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fe0c4d4a18c941588db524a071c464be : SUCCESS in 8m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ea6f732fbb2241f986c3f493c2cad8eb : SUCCESS in 4m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/88d5e9f21677494e85288aff14b96f85 : SUCCESS in 2m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cc0db70b071d4a65b8b11541a856a67a : SUCCESS in 5m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bf82d37b1df14714a1090f152a67acfa : SUCCESS in 3m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/495a218433a244738fde8777bdc6d621 : SUCCESS in 2m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/037a27ffcbc14440875d45eec7d67c33 : FAILURE in 5m 27s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9c4a532be11145039cccfd51ee85ec25 : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/543d5e39b3f74ffc893e4fa5a69b2614 : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d92ec352361d41bab50620ea05ed8c91 : SUCCESS in 7m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/56a01e3680e0443bb3cdbc356135fa17 : SUCCESS in 17m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d26ba60f4f543bdaff36ebca9a7aa2c : SUCCESS in 5m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3b12d8a8553b495282e995d1a7150262 : FAILURE in 7m 36s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8d093a006549405bada86052e24e8c6e : SUCCESS in 10m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cb8b421b28544fe7a49ce269391cf077 : SUCCESS in 1m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/66684f19689543bd8e5176da8f073a17 : SUCCESS in 10m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2dfffcbd69af491597bacd8c5fcae6f6 : SUCCESS in 5m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c9004eb6dcde4ecfb28542abc7222d3a : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/31e0b8ba5dd8468f9f8e29eaeb74e5b5 : SUCCESS in 11m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bfedfd25c7fc4c1f902deabbb1d6d739 : SUCCESS in 23m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b9886f4216c44cabf55bd0667f198d9 : SUCCESS in 4m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cee95a491874450bb144dc36abb3e39c : SUCCESS in 13m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6c4f039e6d584eac985d3f77d3399658 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55c6766ee57e4f128c4018da6f804aa9 : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/626ab3a6c8294798bb5f3f4a4c98e512 : SUCCESS in 31m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6bbdfaee4e584b4c949d9e8ba2eae507 : SUCCESS in 6m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/87ae087b95cb45d6abc26253dbe1b84d : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8214ac63e2b54734829961ada846f39b : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/adf9103404ac4a8f871da9034538e2e0 : SUCCESS in 12m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/42f52719236149f8938988f463752a62 : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/210d122b87bc4b5baf112bffadf48767 : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a9399eeab854e76972788d985282417 : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6ccdb31cd089c04fc6b3d2b662a2e4021331ba1b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 20:56:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b72ed32a065141040d6bb91de123238c7068c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c986dd32c60867b627ab71f344f8062534eae6c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1eb6d4e868a41b49da8e390778b1947\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78445a702d50417da69b59e65041e01a : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc03aa7cccea4c97bdb8de221b0603ca : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1246f70da191409cbba9779a9c28a071 : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba3965050dff43a29b564a6dd44517da : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76c9eb73df9d45cdb8c61c237a711d3d : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a60edd9275a5491585f1700c8023e8c1 : SUCCESS in 3m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/80b6d6d8f571455bb9d0416429d86a89 : SUCCESS in 4m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b73bb4a68a184039919390917c3a2134 : SUCCESS in 10m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a287ccc86fda46ea8c5d6599b9aac963 : SUCCESS in 6m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0c03c0a686ca4da096e577be8aa92e9b : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d24b248c0d2443cfb1c6a8697dc0a4a4 : SUCCESS in 7m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/96ceac8a01b74214bd126d8650b1c176 : SUCCESS in 8m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/764df79916af49b18bb9f0b0df8a3ee0 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99fe42b3902c49f5a80e613c3b291f9e : SUCCESS in 4m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d51c6ea12e834878afd1415d3879f15f : SUCCESS in 7m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/69d67f56a0d442ddb73665f2f01b2412 : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aee8c6581e1b435dbcd53709e3cce208 : SUCCESS in 11m 07s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2e4ae09626ac4dcb86ed090267b87b08 : SUCCESS in 21m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cff51101275d4c54b81faa5436d1cc62 : SUCCESS in 8m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b929f663262f4718a469fe7880949fa3 : SUCCESS in 13m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f143e8f2707a4c8a9e7b64d18495f92b : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/093b2664f78842149afd60b82e26b8ee : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ace8e0b81194669b9140e3645fd73bc : SUCCESS in 21m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d87af4cb85e845fa82feea9f65edd7ff : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe24e70547b04fac87326ebf01b347a7 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76bce2abd30d4653867b3534b6527e56 : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/daedeee52f834398a1f8cc640cda3925 : SUCCESS in 22m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f5351e9118ee4257a4b872da1717f5a5 : SUCCESS in 6m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/105b5272b30e4e17a8104cf174d02f44 : SUCCESS in 32m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00e411b6fc204e209c238dc3153864ae : SUCCESS in 1h 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"779fe09298947954408c1fb601785ab8601bcbdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:48:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62556db1eee92c78f12e5095d7e3d67a7189afbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:49:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eade73da5e4b438d8b4b255f8a6e71e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f08129107646989cccbb5e745f8185 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f273fa0eb2414d1e4c2f019802c0f16839719c55","revisions":{"f273fa0eb2414d1e4c2f019802c0f16839719c55":{"kind":"REWORK","_number":1,"created":"2026-02-13 19:55:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/976811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/976811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/976811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/976811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/976811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/976811/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 19:45:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-13 19:45:58.000000000","tz":0},"subject":"update constraint for tooz to new release 8.0.0","message":"update constraint for tooz to new release 8.0.0\n\nmeta: version: 8.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I165e0af29b3a037228e034e2c3b21abbb60d70fe\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifeae726aad5171192c45aecafe198715b416c60c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f273fa0eb2414d1e4c2f019802c0f16839719c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f273fa0eb2414d1e4c2f019802c0f16839719c55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
