)]}'
{"id":"openstack%2Frequirements~976833","triplet_id":"openstack%2Frequirements~master~I790078bf3aeaef2a1215d4ccaf9a3b9f8c434a92","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I790078bf3aeaef2a1215d4ccaf9a3b9f8c434a92","subject":"update constraint for XStatic-objectpath to new release 1.2.1.1","status":"ABANDONED","created":"2026-02-14 13:28:06.000000000","updated":"2026-02-16 16:22:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a22c9be16f43ed9becf687f74fd0f94f0207b271","_number":976833,"virtual_id_number":976833,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-14 15:08:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-14 15:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9c0e333e10baf8f27c20242c969dafce9a7c8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-14 13:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76ff26f41421dd47dc1d19048ee76e7a4310bc36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 15:08:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb806b30c2e44357b6d5e476e95052be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324239685e324e458ffa28048cbd6156 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774555360ac1493aa6542f4aa80b7578 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b58a990f8ead48958c1a57b3649c93b7 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65e495cad4934796afd56e0531171a4c : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/896d5c6012454de9b2b0b8c0e3c58b92 : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12fdf871f1d24812a530a3f9fba9f7ee : SUCCESS in 2m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/654f9ba9b1f647a68899b0405cdccee2 : SUCCESS in 5m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/16561610321b4f50bd3aa64c4d8d7ce6 : FAILURE in 4m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/605f604fa205407d99fd825d013b9e57 : FAILURE in 4m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1028c8313793408197508733745c019e : FAILURE in 2m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/035907a05a5844ef9ea6d9b97c41b97f : SUCCESS in 2m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fe7fd636eef84494a1015bbb03bebc86 : SUCCESS in 4m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c7debe42358543d9a7b04b3b99550d95 : SUCCESS in 5m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1a0e6050ed024e47a797a26b7727651a : FAILURE in 5m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a6cb1555f55142f7bedb3808df3e500e : SUCCESS in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d3de185016e4968a9d5296eb90ed452 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6ca08aea911f4438ace17f23979a8524 : SUCCESS in 3m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/de6ed0ff7b41464c8f9e4d7105d53f02 : FAILURE in 2m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39e95506254f4493b21838459763f991 : FAILURE in 4m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/89ade1f0515648bf9e78512f277b66fa : FAILURE in 10m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d845bde7243744f58c8498ad1a4f3f30 : SUCCESS in 13m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/090940f370d34040a18352d654d0e03a : SUCCESS in 4m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/39ad9f7076074f36abe9d4d85c119622 : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/88cc286db2034802a053630d4ad28a47 : SUCCESS in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c48b0867879a4f1786fc001b24091a8b : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d5a765d8ee754b7691ca113546dfa4c3 : SUCCESS in 11m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b43e73d634ad4e04a3b3153b6cf6b234 : SUCCESS in 22m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6570f12f6cf54bc18637867958cb73ef : SUCCESS in 5m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/67f4dfd05d94414a98a27bf2128fd736 : SUCCESS in 16m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/91f057d879b84d5691df39ab26bff739 : SUCCESS in 8m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01904c3818ad44c1b7cfbea86285680d : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/247e1b90b61b413687eef793e5778d27 : SUCCESS in 25m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/125368a9b9994bcb8b0063876703a120 : SUCCESS in 6m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/688c13f19bb849cfb2e7144dc3d60ab1 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a70fd955eb247dda8e4523c3d30af9f : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8de6fec8ea2f483889b68dc61842a9bb : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0df6ab865ed646778a20c4a9d4875d47 : SUCCESS in 6m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fca259270d744c25a108d18dd7950078 : SUCCESS in 49m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0972f8cd50aa4101b70d959da2cb0226 : SUCCESS in 1h 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a22c9be16f43ed9becf687f74fd0f94f0207b271","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c96df66d5123fe0fb93ffbd4856eb1c01081215a","revisions":{"c96df66d5123fe0fb93ffbd4856eb1c01081215a":{"kind":"REWORK","_number":1,"created":"2026-02-14 13:28:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/976833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/976833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/976833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/976833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/976833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/976833/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:18:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:18:05.000000000","tz":0},"subject":"update constraint for XStatic-objectpath to new release 1.2.1.1","message":"update constraint for XStatic-objectpath to new release 1.2.1.1\n\nmeta: version: 1.2.1.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I2d4604f8160c2c56d487ed67447f9530d11a0286\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I790078bf3aeaef2a1215d4ccaf9a3b9f8c434a92\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96df66d5123fe0fb93ffbd4856eb1c01081215a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96df66d5123fe0fb93ffbd4856eb1c01081215a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
