)]}'
{"id":"openstack%2Frequirements~976834","triplet_id":"openstack%2Frequirements~master~Ic72cb40ee0ea005c56f30589848f7f771cbabc41","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic72cb40ee0ea005c56f30589848f7f771cbabc41","subject":"update constraint for XStatic-tv4 to new release 1.2.7.1","status":"ABANDONED","created":"2026-02-14 13:32:34.000000000","updated":"2026-02-16 16:23:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d162b10b6d1ff1ff25e80c2922efc9a7ac87c0d","_number":976834,"virtual_id_number":976834,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-14 14:47:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-14 14:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25adf38fafb249f991f4d7f042824864391cbfba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-14 13:32:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea918a825fee994e9f5336029317f0c24c79d355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 14:47:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3640c00fe73b4b4ea5f063ca37e28f70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5b44eab7a444aab0f732a711e315f8 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e0b9755ba845418400bb04d4b55415 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa0d4a9181a44aa48f586aebfe8ea91b : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed62981032744d04abe5be3dbf92b3d2 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4afefc1b93e4593a5566e43f18b0717 : SUCCESS in 2m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57be4c6ee79a4624be66dcc12a682f19 : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae4c3bec4dfd4ce6bee7c937196171d4 : SUCCESS in 2m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2b6d39ec16454b52bb807395375cd134 : FAILURE in 2m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5604cefb5ae04e7ebd32147bd35bb100 : FAILURE in 2m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/187f69835bd64feaae22ef13defeeabe : FAILURE in 3m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/92eb0136ecee4d3490875e9694ea1b23 : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a213f576508840deace6a6fed0e5652b : SUCCESS in 2m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a6341ed885a94b7fba37eb71003e613f : SUCCESS in 3m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/90f1b92900f5460d995a90f4c18327ff : FAILURE in 5m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ce1033e2427e41faaa09a68da02c8847 : SUCCESS in 4m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/32e1e079081c4558b145bbb02b6ed619 : SUCCESS in 5m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/892212ee30764aa4a4a6721f8557857a : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/108f9d0646944ba1a58942e327e55fc6 : FAILURE in 2m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccfe32a6012c46c4aec0b0d23c3aa5af : FAILURE in 5m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db9735befa86433db24068bc8e36f66c : FAILURE in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/11dcccd40e4d4d9ea94401bfc38fb7e7 : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dbebd46a726249e194371b1b86f8238a : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1c12b63b71cd4fcaab1f470faf286069 : SUCCESS in 9m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3b792370871a4e90b6243194d60ed7e1 : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2cb5c7b0382140fcbe8b538cb40421f4 : SUCCESS in 2m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b87f0c987f9c4c77ba12500c8e47f05a : SUCCESS in 11m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/73340a0f47854bcbaa159e0528a0e5c3 : SUCCESS in 22m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/26d181478c2043e79456b78188f363b8 : SUCCESS in 7m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/87e4e0e2a605425db8da655f1f8234d4 : SUCCESS in 9m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/731758447ef4484f8e0cb5db498df7f5 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cc548c50a9546a5a211f75b3acc00bf : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4136ca9b285743fd9fec5f58c32b8503 : SUCCESS in 22m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/431dcfbf462244a38107492f915a275e : SUCCESS in 7m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bac9e1c381b44491a8f5750388596af6 : SUCCESS in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/917b26ca5c434c55be1af5048f95ee02 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e418ba63f4b94eaca6f9b44c406630b5 : SUCCESS in 10m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3c984e5e775c49ca8563c6a30f7c923f : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9bb1c0cc53340f594f1d35d1d687aa5 : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50fa4ef993004660aa980c233fb16ae5 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9d162b10b6d1ff1ff25e80c2922efc9a7ac87c0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:23:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6412b675d3ec7073fccdd25b46114224dfb599f","revisions":{"e6412b675d3ec7073fccdd25b46114224dfb599f":{"kind":"REWORK","_number":1,"created":"2026-02-14 13:32:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/976834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/976834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/976834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/976834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/976834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/976834/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:22:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:22:32.000000000","tz":0},"subject":"update constraint for XStatic-tv4 to new release 1.2.7.1","message":"update constraint for XStatic-tv4 to new release 1.2.7.1\n\nmeta: version: 1.2.7.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I2d4604f8160c2c56d487ed67447f9530d11a0286\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Ic72cb40ee0ea005c56f30589848f7f771cbabc41\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6412b675d3ec7073fccdd25b46114224dfb599f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6412b675d3ec7073fccdd25b46114224dfb599f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
