)]}'
{"id":"openstack%2Frequirements~976835","triplet_id":"openstack%2Frequirements~master~I8cc6bc5d7ad32e047853b416ade65f4bdf50c5fd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8cc6bc5d7ad32e047853b416ade65f4bdf50c5fd","subject":"update constraint for XStatic-Angular-Schema-Form to new release 0.8.13.1","status":"ABANDONED","created":"2026-02-14 13:32:43.000000000","updated":"2026-02-16 16:22:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7678bc6835fd3ecae9a52e0287958341216b7c5","_number":976835,"virtual_id_number":976835,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-14 15:27:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-14 15:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dcf99c2d1c43e646322f2e717c326150276cfd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-14 13:32:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d0686915aa0325025c77b9e751a8291b673244a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 15:27:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd4271de49db40a49731f1e35ed891c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11863b9c0a1843d4906afc1da097acf3 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72a28278f9947b988cd26a3dd8cfb43 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4151deba3d52415bba501a13513d840a : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c822df887bb4000bd515fffe70f4d4b : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/797ee492e6f04e409ba39d0e607b3fea : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b6ed705226240fa8c34d60c889bf52b : SUCCESS in 2m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ebec80471f24eb1b12a0037ab5c180a : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1605d7af891c4f23956ec8f296153b44 : FAILURE in 2m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ce37fa8597c6420a846156dd552b8962 : FAILURE in 3m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/83f0b80bc5104f738f50f9f30e89f092 : FAILURE in 3m 36s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/13aac673864146c68d376f47f6ecccdb : SUCCESS in 2m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/eccff0ba70f34579a6fdf9e28088280b : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ca1f03179c144d5bab88b27a17072694 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/529bbffb69ec45cba2fdd1cb32ee79e6 : FAILURE in 3m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/be03395f5bd74ce3926c98cbaeb7ae54 : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/720591e1d77e49a386b8075fd0ee2452 : SUCCESS in 6m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9d63a2a15ca742da96f45b9283368e06 : SUCCESS in 7m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d1e5ee8edeaf44de8336d421e5585f33 : FAILURE in 3m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddf53d8fff6541e1b49a111cc517466a : FAILURE in 2m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f2763fca79194da99e5ed297d17cd0a7 : FAILURE in 7m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6407d84f7bc841179227bbc24938ab40 : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cfaa384d33434b479801196ad8b8b488 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9e50d7c9efac4cb8849e56a8665ef06d : SUCCESS in 8m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c27c4430a5ec42d3ab48f2e3079cbc6c : SUCCESS in 2m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9bb570ca822a40a6ad7efbe6c38cc25b : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c498575ed4eb478498587ce13da0157b : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a319e68f0ec540da8c0bb7feac1b4ab9 : SUCCESS in 28m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f84963beadab4fb0a652e2b490a7c94a : SUCCESS in 3m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9e6e6df27a684dd097a5c9f9248faa6c : SUCCESS in 12m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e8ca4d41bd5743a98bc25d6171cfa1fe : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7493acf7a325417e85c6ec201e69a0aa : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4c11b3bf0a148e7a99a1b2cdf083b64 : SUCCESS in 28m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/78a8e1a554744748be05a809be2f4440 : SUCCESS in 5m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/88a82189fff24f8d80a6fde0622955e2 : SUCCESS in 3m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6636ba614144e8db74866de91a1172e : SUCCESS in 6m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ae82d3da749a45e4989221dd80f158c2 : SUCCESS in 6m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4bc88772714a49a3bfa68cde562e3e18 : SUCCESS in 6m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/355928cce98547f2a4cbfee13472c7fa : SUCCESS in 43m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8e6790e95274b9ebc3e1fada9271c02 : SUCCESS in 1h 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a7678bc6835fd3ecae9a52e0287958341216b7c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 16:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"def1273f285d7f9aac3387af08bde7ad6c973dcc","revisions":{"def1273f285d7f9aac3387af08bde7ad6c973dcc":{"kind":"REWORK","_number":1,"created":"2026-02-14 13:32:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/976835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/976835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/976835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/976835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/976835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/976835/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:22:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-14 13:22:41.000000000","tz":0},"subject":"update constraint for XStatic-Angular-Schema-Form to new release 0.8.13.1","message":"update constraint for XStatic-Angular-Schema-Form to new release 0.8.13.1\n\nmeta: version: 0.8.13.1\nmeta: team: horizon\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I2d4604f8160c2c56d487ed67447f9530d11a0286\nmeta: release:Code-Review+1: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I8cc6bc5d7ad32e047853b416ade65f4bdf50c5fd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/def1273f285d7f9aac3387af08bde7ad6c973dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/def1273f285d7f9aac3387af08bde7ad6c973dcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
