)]}'
{"id":"openstack%2Frequirements~976860","triplet_id":"openstack%2Frequirements~master~I4ba9fa8789be507a428cd13cb6fc9e7a60c6fce6","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-15 18:33:25.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-02-16 14:14:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ba9fa8789be507a428cd13cb6fc9e7a60c6fce6","subject":"Drop vitrage related xstatic libs and fix CI","status":"ABANDONED","created":"2026-02-15 09:49:14.000000000","updated":"2026-02-16 14:14:13.000000000","total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"2df004d764583f4931b62e57c95f28a2c60af265","_number":976860,"virtual_id_number":976860,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-15 15:35:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-02-15 16:39:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-02-15 18:33:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-15 11:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-15 12:17:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-02-15 16:39:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"65cfb813c4d33439c444f7359b2d6e8c300672b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-15 09:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c76c74c431ff8076e1cd4d4afe2b053ed0d449af","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-15 09:50:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7543465111ce9b336fb3a07405a4df4514cd57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 11:39:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9db21bcc35bf4960a87d48d257e0ffe8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29783181b8b4b7fb6b05dce8febabf3 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d2f117715c435984b35c06e81539b2 : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca1374f0ff6047aabd73f960c893d736 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3ffceb9f9d84212a4f190b5be294d70 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2efc6e10282e47bbae7d1bfb3eb00aff : FAILURE in 1m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83fa8a2743d841bc8388d1245231b319 : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/269bad2087d34171b863cf5c762e8c5e : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ab273449b7a44e80926f994d3d084d67 : SUCCESS in 5m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0818d03516744ce98689ac7b1ca01691 : SUCCESS in 11m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ff096d501a32466a84bd2feab95dbe7b : SUCCESS in 6m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c58afc1a5ce34440868d3d67a1f5ecb0 : SUCCESS in 4m 21s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c9cef5df6d114de5b650a8fa464efa15 : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bb2faad430d34671b925738cd9043a10 : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bedf997ebcce439fa7ac8fa0dc48e960 : FAILURE in 6m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aa4f855d19784149a9e61c8a037253d4 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e436703d40f54fd0a7861ca92f0e30d6 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/005d0a6a432b483eb8aaffd72304b033 : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ce9f3acd61c4e03a9dca9fe46172b7e : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99c9db7712a14ba2aa2c68de799b296b : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f54fcc259b9d4f5bafdab006ef6133e8 : FAILURE in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/42b332cafdb9419b98af1727f036a4d9 : SUCCESS in 8m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d4e37450a8894fba845c50533d68e5bd : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/71315a63b72d4114bd2e83f1360f2369 : SUCCESS in 9m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6b41032e660f4fc29cd937651ba5bc66 : SUCCESS in 6m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/933beaf6f08e4b248d8011bbe98d0ca8 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1252f2adbf7146539b2c4d068ddd3ee3 : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/327c3f427e8e47d2baf1a6fde1b46b0c : SUCCESS in 27m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/681e925378484f288f6085174a37e8d3 : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5fb4187a21e34116869b5278ae9b370c : SUCCESS in 12m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7b790b9ec2324a788af1f2818479e1f4 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8a8351f4cc3465e80a3d5f4888f1959 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5644357b8dde43fb9cecb5263d46c153 : SUCCESS in 24m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/59a616fdcc554ad18d16d7439a60ed82 : SUCCESS in 6m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b78dafe8e25d4596ae4c71c99723c329 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/651a92fe58064257a5d4ba6ab9e8149a : SUCCESS in 6m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c4adf7859f914a44b177abd5b68df27d : SUCCESS in 19m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e7c4dc653eb74c369c334e4601aa59c9 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f870de59713c4cba945dc52a95657f01 : SUCCESS in 36m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e63daafc09bc4545aa2a610d4775207e : SUCCESS in 1h 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2fe4a002111b34248692305dcea59bcc3242cafc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-15 12:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"037c16df2b026d03fde2abde039cc33bd7adaac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-15 13:46:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b18f848ca10c1f248f116ecb5b211331d942c2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 15:35:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ecff21d87b4b169b5bcca015911d56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c1d4e39665437a829cd3d8c6aee61b : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58d74d5ef34435bbd5b2d4fa6ce0a5a : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac19cebae909472f86a18311db53b237 : SUCCESS in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b07045945a3c4b868a198447f0d0fa5d : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6521dae385540899adef32a5ed486b6 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf0cbb991d944d9fae3bf55db45d31fa : SUCCESS in 3m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae1f2b27379c4beea403f142a1496a86 : SUCCESS in 10m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a56b39201478420a8dba4e019cad3330 : SUCCESS in 3m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/45af8213c3324d7f8e441eea0a35d108 : SUCCESS in 7m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6fcd675d798346458adc821dac91c702 : SUCCESS in 5m 43s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3d074cbf43994444bec805b198a92092 : SUCCESS in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0132d9e9e5654017b471ba5d63eeb24a : SUCCESS in 7m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e17974025bf04679bdf579e27b118e40 : SUCCESS in 3m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f832c31720e54611bd16cbd0bf86b4b3 : FAILURE in 5m 08s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7d99f023230345a1b6e721b65fbddc37 : SUCCESS in 6m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a85815b39c184b0ab200235fee1644c1 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c57826f634944be09acc5b30ae20391e : SUCCESS in 7m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a080fcc7e5974e779ce9db59ab9488cb : SUCCESS in 18m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b55f1a81f81548f89fa55e89c283247e : SUCCESS in 4m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c6ab9cc8977a45e79020273e8ef6afec : FAILURE in 8m 22s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e9aa7471a86a4788aacbbcfe7aa0e05b : SUCCESS in 12m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fe267cd7e053452c89296995ade3e1e8 : SUCCESS in 4m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6c3a6221fd1e4b409ee0a122ab8cf76f : SUCCESS in 7m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/81324d23510347c99b58f05108c95ca9 : SUCCESS in 6m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/28d17b77c4b94461baec09a3e3de8e8e : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/82b51de22ca746d3bce1a226170f0139 : SUCCESS in 9m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9ef2b8014e53468290d295daaac11697 : SUCCESS in 27m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d1941cc64719460db13ed9f41bf8e474 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/231dd2ad4930408f8ef46d3c63a3457a : SUCCESS in 13m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5e289ceb7480435a99ff93c3ca10a6ce : SUCCESS in 8m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5417cac301047a7953f65e627031797 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/62b894d02f834522a45fe3ea093b4f42 : SUCCESS in 20m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d8887e3a1b4847b0bf1a09ae2dd723e1 : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3047a4e2a104474bacb715799e0e0194 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8505428dde6f423fa3449b1fa43ebee9 : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/532ef198d15348568edaa24fb1decee2 : SUCCESS in 18m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5c7c63dce2fc418b89b28f4953087a4c : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0d5aa54d8dd457789813b01ced52aba : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf38bb4bdd4d4b85ace178daf3443359 : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0f5c71b200bcfec67c1453683d9160aa046568ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-15 15:43:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25210948b887f1e106e6fafb536c3050fbb852a8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-15 16:39:36.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7134fb74cc942c2216ca0d9b7ddb4ccd4926c9fb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-15 18:33:25.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2df004d764583f4931b62e57c95f28a2c60af265","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 14:14:13.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e049fa0ee2fd9650aa00257e5db9c50ac5558c8","revisions":{"f18604c95b2570c1781990a3465d42b346856ce4":{"kind":"REWORK","_number":1,"created":"2026-02-15 09:49:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/60/976860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/976860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/976860/1"}}},"commit":{"parents":[{"commit":"4ab0e5af22ba05e1784a2a9255b0c8b077d3cd68","subject":"update constraints for various XStatic-* releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ab0e5af22ba05e1784a2a9255b0c8b077d3cd68"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-15 09:47:49.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-15 09:47:49.000000000","tz":60},"subject":"Drop vitrage related xstatic libs","message":"Drop vitrage related xstatic libs\n\nThe vitrage project is inactive, so we cannot properly update these to\nfix the breakage caused by the latest setuptools version.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I4ba9fa8789be507a428cd13cb6fc9e7a60c6fce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f18604c95b2570c1781990a3465d42b346856ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f18604c95b2570c1781990a3465d42b346856ce4"}]},"branch":"refs/heads/master"},"4e049fa0ee2fd9650aa00257e5db9c50ac5558c8":{"kind":"REWORK","_number":2,"created":"2026-02-15 13:46:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/60/976860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/976860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/976860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/976860/2"}}},"commit":{"parents":[{"commit":"4ab0e5af22ba05e1784a2a9255b0c8b077d3cd68","subject":"update constraints for various XStatic-* releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ab0e5af22ba05e1784a2a9255b0c8b077d3cd68"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-15 09:47:49.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-15 13:45:35.000000000","tz":60},"subject":"Drop vitrage related xstatic libs and fix CI","message":"Drop vitrage related xstatic libs and fix CI\n\nThe vitrage project is inactive, so we cannot properly update these to\nfix the breakage caused by the latest setuptools version.\n\nAlso make the cinder and ironic jobs non-voting as they are currently\npermanently failing.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I4ba9fa8789be507a428cd13cb6fc9e7a60c6fce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e049fa0ee2fd9650aa00257e5db9c50ac5558c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e049fa0ee2fd9650aa00257e5db9c50ac5558c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
