)]}'
{"id":"openstack%2Frequirements~976903","triplet_id":"openstack%2Frequirements~stable%2F2025.1~Ib97eea1037fac541c8ebadc526ae9c3bb160f685","project":"openstack/requirements","branch":"stable/2025.1","topic":"pin-setuptools","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-17 13:23:55.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-17 15:08:42.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-02-17 11:53:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-02-17 17:44:02.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-17 17:44:02.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-17 17:44:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib97eea1037fac541c8ebadc526ae9c3bb160f685","subject":"[stable-only] remove pkg_resouces from reqiurements-check","status":"MERGED","created":"2026-02-16 08:51:30.000000000","updated":"2026-02-17 17:44:02.000000000","submitted":"2026-02-17 17:44:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976903-pin-setuptools","cherry_pick_of_change":976259,"cherry_pick_of_patch_set":3,"meta_rev_id":"5e9f29f6a89d21059721135a4fce138632f80027","_number":976903,"virtual_id_number":976903,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 17:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-02-17 13:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-02-17 15:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-02-17 11:45:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-02-17 15:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 08:51:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-16 08:52:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-02-16 09:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 11:45:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-16 13:39:53.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-02-16 14:04:30.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-16 14:06:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-02-16 14:24:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-02-16 17:17:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-02-17 08:51:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-17 09:26:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-17 11:45:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-02-17 11:53:17.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"da968c444bbfc301ea949e3e9a2d29216349d853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 08:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9c5abac19009fe01a232c016e7c803304ea8dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 08:52:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ae524429fdc017322f38f6f8aa581bb3a4c151a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 09:58:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/957b7ceb9e72482ba8e4f45c8099b2c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86b6694f9a8146f2b6b0d87409cf9ad2 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a07df5ec1a40298778597d38eb6d09 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32b00714f2624468a553c3dbbf858d80 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7751503f2b8472c9d2881d5326d686f : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5086a47d63634f7dbb40c1a71766b877 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57d4582d78d94091999733ef7fe35d2a : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0893a8a65e6d4ae5a7b8cc6ec2b7dfd7 : SUCCESS in 1h 01m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c3c408ebd66449c8b5e248abf44d4d9 : FAILURE in 42m 07s","accounts_in_message":[],"_revision_number":2},{"id":"90331668a9d9ec2af4e0e93b091fb5d3d95ca06d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 10:54:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb246de0b0ad5cdcf6098df68024d986ad18326c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 11:45:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e2e1968f9e9893fda2631ac298105a28f5eefcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 12:50:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a620312468468399faa50c1f39e8f9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f87c988f8ef4da1afdf2db547ce8743 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c345875b114a98ad1b570418a74c8e : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c30f751850114ea88968c626f0816fda : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4fbc58961ae40a9840bdf75815dc189 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47acf5675c8547539f962c3143d878ea : SUCCESS in 2m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f920575b25e24e0f8f9e8f0adf377952 : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55ba66f280f04fee8bd09f6f48aab9b3 : SUCCESS in 1h 09m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0413b79e1ede4a16a3273ed29cacd311 : SUCCESS in 1h 49m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b95765fac77f429fe5babf1ac36d123f9299cfde","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 13:38:22.000000000","message":"Patch Set 3: Cherry Picked from branch stable/2025.2.\n\nThe following files contain Git conflicts:\n* global-requirements.txt\n* tox.ini\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ef2387045869506f46d6618ec9af52ac129bb035","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 13:39:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69185ae7d51ff83f828c010a24ad63d46b6e931d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 13:40:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d86419766aeb4f9534d3166c151fd5baa5e37f99","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 13:46:57.000000000","message":"Topic pin-setuptools-stable/2025.1 removed","accounts_in_message":[],"_revision_number":3},{"id":"04a95da715e5c372f0c49b67a42ca45828c20947","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 13:47:10.000000000","message":"Topic set to pin-setuptools","accounts_in_message":[],"_revision_number":3},{"id":"2559ab5d60bf62b43baae233fc1180d77443a4e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 14:04:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28b6e083b0a92af8d97f874b4683d984df2e7c9a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 14:06:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1d233628069eabca6305b84f5f8021dc0c84c037","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 14:23:55.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"e1e1fbaa03b81681c44cbac33801ee4b94807011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 14:24:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"09871bd118d83ceaa9cf39b2bc5334e4e4ae23e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 14:24:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a1355993f070317521fa69cf5b45a339dcf2f17e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 14:30:32.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8674aa1c75db78f65aab3a53e1f467e8706fac73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 15:33:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/673dc6336bc84a1ca6a9ab3595975af5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e7adca7e54e849099d6da751dfa00a79 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828f06d6473847ebb43aa2747d105965 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9900c1cc5ae645bca946921c26946a29 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12375384d9984ff0a3daf631709f7bc2 : SUCCESS in 2m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53b44f9918f84e3c91fdff1f3c1619fd : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50d5e24186e74d24837da5f2c41c63b4 : SUCCESS in 3m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59a836082c64028bb580282c29e4eb1 : SUCCESS in 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41063401a84645a9989c49cfc7cf2500 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":5},{"id":"c3b4aef9dc178bf6e66526b8512b7e9d67648050","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 15:34:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ad9fd23ea631c291864a303d87a2e4b78640737","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-16 15:46:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb1d2b9558ac0b6ea320818e368e1064ed309007","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 17:17:14.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"59acb4f2cc663a81ea2f3a57a4a18bffbd20494f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-17 08:51:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d0971a6b6b064eb1c4b5875cffc3db8e6fa9d619","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-17 08:52:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5c20b8003357ccbfa996b464fffc7d98fad5f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 10:02:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a48cfd511cf46ae81fd120cb0ad2764\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/115520ec7e8a4c33bdc708c07c56b36e : SUCCESS in 1m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be2d8b6018b481f827990f175d80270 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb457688fe03454b9975d13fe465f091 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c99cd8b95aa4e8f9ef939e167a6f570 : SUCCESS in 2m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e85cb186b9dc4a64a3c26bb9aade8ec1 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/886bcbed325540ec8885271936b8e0a3 : SUCCESS in 2m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/271525fbcadc4f829dd71833fd6908ff : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/851a2eadf895421d8e80f840cd889e56 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":6},{"id":"d64398aa44346c3d230a131fbff12fec20cb17e8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-17 11:45:02.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a810ca6e14b225f7d755a32a2de046bd3dbcb40d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-17 11:51:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"044f7ead5a523d8de5d19ddc3ef451f1330df2af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-17 13:23:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9acbc1e286cb4c88ea3a13d136ba3b2893d5a2a8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-17 15:08:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7fec20c1f4e98d928d34abf1bb03737081ac883e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:09:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0501a04322b8b9f40d5af910160a6b9ba4e0f4ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 17:44:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd6d976338841b08f3c61a508e80f91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b510bc6da5847d0acb382e3b461ade2 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883af49005ca4e5d99de1381d08ea687 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a9a7e80e774849ae31619377817ef8 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1473118b9a0949e58f96e239bb002267 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9e42c11105f4e67b3c5c2f202ae8d27 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac8441bc027e489083670a144a8d1a52 : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08eaa50a52ea41cc8480cecfb90fdf8a : SUCCESS in 37m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4c03ce6b63e45598f49ba43f5c37887 : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":6},{"id":"5e9f29f6a89d21059721135a4fce138632f80027","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 17:44:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","revisions":{"1be5933209f72d6050834f993d3fb5232cc124c2":{"kind":"REWORK","_number":1,"created":"2026-02-16 08:51:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/976903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/1"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 08:51:28.000000000","tz":60},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 2 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976105\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be5933209f72d6050834f993d3fb5232cc124c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be5933209f72d6050834f993d3fb5232cc124c2"}]},"branch":"refs/heads/stable/2025.1"},"bd0f4de4be8647ecb036f4c773e85bef47430676":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-16 08:52:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/976903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/2"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 08:52:22.000000000","tz":60},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 2 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976106\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd0f4de4be8647ecb036f4c773e85bef47430676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd0f4de4be8647ecb036f4c773e85bef47430676"}]},"branch":"refs/heads/stable/2025.1"},"7dbdc1bdb7d0890bbf633834452403926709c611":{"kind":"REWORK","_number":3,"created":"2026-02-16 13:38:22.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/03/976903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/3"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2026-02-16 13:38:22.000000000","tz":0},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 2 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976105\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7dbdc1bdb7d0890bbf633834452403926709c611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7dbdc1bdb7d0890bbf633834452403926709c611"}]},"branch":"refs/heads/stable/2025.1"},"01174f2d2c96b170ac19303eee75a1f5155da441":{"kind":"REWORK","_number":4,"created":"2026-02-16 14:04:30.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/03/976903/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/4"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-02-16 14:02:36.000000000","tz":0},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 2 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976105\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01174f2d2c96b170ac19303eee75a1f5155da441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01174f2d2c96b170ac19303eee75a1f5155da441"}]},"branch":"refs/heads/stable/2025.1"},"bdd7a3922d7f7804fd1799fd778466f49d456a2e":{"kind":"REWORK","_number":5,"created":"2026-02-16 14:24:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/976903/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/5"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 14:24:12.000000000","tz":60},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 2 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976106\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdd7a3922d7f7804fd1799fd778466f49d456a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdd7a3922d7f7804fd1799fd778466f49d456a2e"}]},"branch":"refs/heads/stable/2025.1"},"7b62011174e030d696d7ca42eb11ad82f2b19a3d":{"kind":"REWORK","_number":6,"created":"2026-02-17 08:51:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/976903/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/976903/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/976903/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/976903/6"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 15:44:45.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-17 08:50:38.000000000","tz":60},"subject":"[stable-only] remove pkg_resouces from reqiurements-check","message":"[stable-only] remove pkg_resouces from reqiurements-check\n\nthis is a squash of 3 commits and a cap of\nsetuptools to \u003c 82 in setup.py\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/976106\n\nReplace use of distutils.version\n\nThis is gone in Python 3.12. Ordering is slightly different with release\ncandidates now (correctly, IMO) placed ahead of general releases so we\nhave to update a single entry. codesearch.o.o [1] suggests this\nshouldn\u0027t affect anyone.\n\n[1] https://codesearch.opendev.org/?q\u003dpysaml2!%3D4.0.3\n\nChange-Id: I4cde27f048889dd557c0473138bd8a75c1e1cc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 2d88770535d5a985a461625fe6443039932671c3)\n\nRemove check-conflicts script\n\nThis script made sense back before pip had a proper dependency resolver,\nbut since pip 20.3 [1] this is no longer the case. Our jobs to test this\ncan simply install the dependencies and let pip complain if it cannot\nresolve things.\n\n[1] https://discuss.python.org/t/announcement-pip-20-3-release/5948\n\nChange-Id: Ifc30478dfffd52f126e0ad7535468cdd0b1cd36a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 63a179077672327ccd02e34010a74f15e9d9eea9)\n\nAdd pyproject.toml to fix install issues\n\nAfter capping setuptools in setup.py, there are issues when installing\nour library in a venv. Add a pyproject.toml file to avoid this.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ib9b7ecde0210a3b11bcb70e96e295cf73f2e0954\n(cherry picked from commit ad78d8346bd5e3abd4ba5a901867baab6581f0d8)\n\nReplace pkg_resources\n\nThis will be removed after 2025-11-30. Get ahead of that breakage.\n\nChange-Id: Ib97eea1037fac541c8ebadc526ae9c3bb160f685\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 7f008f902d5470093c67d7b03cd3a896712702f0)\n(cherry picked from commit ed0f971305bee6b48fb803e479c62ab0118c15be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
