)]}'
{"id":"openstack%2Frequirements~977046","triplet_id":"openstack%2Frequirements~master~Ia22ce36d22c79efdb975f3a1795307872329aae2","project":"openstack/requirements","branch":"master","topic":"ironic-voting","hashtags":[],"change_id":"Ia22ce36d22c79efdb975f3a1795307872329aae2","subject":"Make cross-ironic job voting again","status":"MERGED","created":"2026-02-17 08:35:29.000000000","updated":"2026-02-17 19:28:10.000000000","submitted":"2026-02-17 19:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977046-ironic-voting","meta_rev_id":"311bf91603d86646314222f612898105c3aad984","_number":977046,"virtual_id_number":977046,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 19:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-17 16:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-17 16:41:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 09:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c24c6ac68b3a0ca33c46d32e70d0c122a6f8d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-17 08:35:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f09f9fe5d62140b4278d73f58db6c61fc56c7682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:33:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e582f37dee6042a9af0917644ee45236\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf95f5eacfc4404b9bb785b6bcae2fe0 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c7cbd0f66341d09f61e7fe60159eb0 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4f1d89cff8d4a3696d06b6e0534a9b7 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cc4ad01f5a842c29866bdc050a9f019 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a6f02e6eac14e86af132c9548e589e2 : SUCCESS in 3m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7f100b8e95c3415eb9d22a99a8af4f64 : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/527434662a144e04a3a5db2128927e12 : SUCCESS in 36m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89152a20ee0f4156b89ea3bf1e89d65e : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ffa9a8bf37a069e2596ab206a534dd9fd2541009","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-17 16:41:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8732623628f3aaf5f1ba39c9d3510220cc88ed74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:42:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbbd46a5bcad4ab8df919d9d0fbb072f72750cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:24:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eadb23c55304dfbab37d424be746bf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fec91b660f4bb18b67f82a1d04002c : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a64c3ad115480ea195dacd842290fa : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc645b0baf5b42cdb88e7345a4eb9b32 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/224cdf4b24fe489eb881b73f56a86bce : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/525d3415f0d84d30ab3177e51f497dd9 : SUCCESS in 3m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d8693d7086a84b3a9557fa2a5dee90b9 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b27c17db34242d0a14b3c56c3e73825 : SUCCESS in 36m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75b07f415518433e9a65bcd1d714bfee : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":1},{"id":"dc57b92b567014438c2f07fcd34b393c8d9a0866","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:24:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"311bf91603d86646314222f612898105c3aad984","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:28:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35166caa17f54e63aab29e71a8b87518\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0433ecd581b04ec1b9165cd37b414f57 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f297d278f2464de6e9a7f4a4c5722a4290bfc1ff","revisions":{"f297d278f2464de6e9a7f4a4c5722a4290bfc1ff":{"kind":"REWORK","_number":1,"created":"2026-02-17 08:35:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/46/977046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/977046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/977046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/977046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/977046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/977046/1"}}},"commit":{"parents":[{"commit":"dcd4a1b8c38e9a0b82bb496dff62ae02bcc952ac","subject":"Merge \"update constraint for taskflow to new release 6.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcd4a1b8c38e9a0b82bb496dff62ae02bcc952ac"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-17 08:34:01.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-17 08:35:12.000000000","tz":60},"subject":"Make cross-ironic job voting again","message":"Make cross-ironic job voting again\n\nThe issue for ironic was fixed in [0], so make the job voting again.\nThis is a partial revert of [1].\n\n[0] https://review.opendev.org/c/openstack/ironic/+/976999\n[1] I39d2c867476106c7127a2365d31713c8d07abb38\n\nChange-Id: Ia22ce36d22c79efdb975f3a1795307872329aae2\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f297d278f2464de6e9a7f4a4c5722a4290bfc1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f297d278f2464de6e9a7f4a4c5722a4290bfc1ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
