)]}'
{"id":"openstack%2Frequirements~977082","triplet_id":"openstack%2Frequirements~master~I263137e527ff83d45e7068e02e5bf2f99bc01591","project":"openstack/requirements","branch":"master","topic":"bump-testtools","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-23 14:35:06.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-23 18:24:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-02-23 13:02:10.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-25 11:25:18.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-25 11:25:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I263137e527ff83d45e7068e02e5bf2f99bc01591","subject":"Update testtools","status":"MERGED","created":"2026-02-17 13:31:16.000000000","updated":"2026-02-25 14:33:20.000000000","submitted":"2026-02-25 11:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977082-bump-testtools","meta_rev_id":"6ea08102e1a0d174a8a9018b77302a966b916597","_number":977082,"virtual_id_number":977082,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 11:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-02-23 14:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-02-23 13:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-02-23 14:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 13:57:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-02-17 14:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-17 16:40:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-19 11:41:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-19 13:07:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-02-19 13:18:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-02-20 07:17:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"35bb3383e9655f2d109e30335aad24d1798d3c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-17 13:31:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40df34a9b99b1d43743cc64f5fbe945cc06976ae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-17 13:57:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5869ef4d407c58b9bffc8d319effea55341ac5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 14:21:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8af58b86dc2c43d2bc26c2b5d81ea5bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad7548dfe6e40bd9796001301a85e3e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79765e03f709499cbdc78d9cbba46efa : SUCCESS in 2m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02dd2b6894584b9db0809c07a9d282d7 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a982e8019cc844d5a1ac6a6b87b37f6f : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cb2aec51db64859a115536d16b77265 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f75d527353c4e189d5a8c2b8655856f : SUCCESS in 2m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/91a29d429daf49738091b2498300fb4d : SUCCESS in 3m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/04ac1c896c614b59ab33a34c6df0ca79 : SUCCESS in 9m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/99ad6c380b6c40eb80fba76fefbc7b11 : SUCCESS in 6m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2161a2f5e9b44efcb0eeb9c0a55d6a17 : SUCCESS in 10m 34s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a7869262be374b3d82d5af1531af2aca : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/19334b2e1cae491f8f29897c1cd22413 : SUCCESS in 4m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b4b5a600a5a94a3cbe3b4c45be8c43dd : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fa6a2aebbd1f4e23aa20802b4dc9f4fe : SUCCESS in 11m 24s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bb42c44ccd074516b7a5a3cece9e5895 : SUCCESS in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a2ede7e583674eff88d75ab155b229b9 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f84ee06c89ef4c2186a212058f7211c8 : SUCCESS in 6m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/71028237008644dcbc66ec9a32e9606e : SUCCESS in 8m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4095c5e268bb42c38bc76042a1585c8c : SUCCESS in 4m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8cd27bbb38fe423dba1d606d6e9e8ac8 : SUCCESS in 5m 18s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b778780c02fc406390ee0e2b72eff9a2 : SUCCESS in 10m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3090329713804133ab833a2608ac87b0 : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fff8ddffe8ff4311b13bc2b9f6b56b77 : SUCCESS in 10m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6ddf0e21a9a949369b23f88d0dd243c1 : SUCCESS in 4m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/058a9464e4b447cd84799a10dbb0be1c : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e98a8d0cec8f4bd39d3e7027a194d31b : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/25aba805871a4f40ab59c96bccb02670 : SUCCESS in 22m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c95ed9e057944b80a8af167ad9ace2fc : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cb996608079b45b3b3a141889757cff3 : FAILURE in 5m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/236fe98065b64b60981e373f9360ba9d : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e267ab32cb294c79b933ff55e9a06957 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27517fe555624a02b33c7baff4ebce86 : SUCCESS in 23m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ec2a6b1c5d7c49f4ad2ab371722947f6 : SUCCESS in 3m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/34685b124f78414a8b180880c89db426 : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3af0ace7cca54cdd99c1891ad913e157 : SUCCESS in 5m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/20e870d9926e4e97a4d5bfe18149c3bd : SUCCESS in 14m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5ec8ef382c5d4a0c9dd8dc56f12327df : SUCCESS in 7m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a007ebefed84cc58b926a4f518c2deb : SUCCESS in 36m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34b3294756284320805086f39b995302 : FAILURE in 49m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8b1765a3d5900d792d6a70ba5ef991175f560341","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-17 16:40:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f6eb5e0f6d26486ffc54258da1eefdcc114bc84","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 11:41:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c21e46e859e15233de813cebe695cf5e0a08b803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-19 13:06:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"179e3ff8ad28671bec137f50db5bc459dbdf72f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-19 13:06:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"391d28bc488730d2cb24e6bd52a7aa307fdec111","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 13:07:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"04d94f3b2217a98f0db9bc7b6aa4b2b23bcc7f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:54:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb2deeb52c4b49f88ac3a6952a43b14e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014c26a4703f4949b866d2b65cacd093 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73462013153b41659b1bf28f6f3ecadd : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/944bf3b56e4f4e278365e02044781eac : POST_FAILURE in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d505d74d45694c38bd3a8914fda0e6a6 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b5537175c9d4aa4920f3ee8f97c0053 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/437dc109b26b433196dfa1a8262de7bd : SUCCESS in 3m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/de5d42be26be478cbfc929d5e948c204 : SUCCESS in 3m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5d10ff8b61614eea8a46b40228301870 : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8c48fecf44344cd4b58900b57abb98f2 : SUCCESS in 6m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/34703f8bf74a4c6db5d789adb64a8c3d : SUCCESS in 5m 29s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2059f9a5a43e4785a07a37525f218386 : SUCCESS in 5m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e1919a620dcd4421ac238b7f02b58de2 : SUCCESS in 3m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/30c7274249ca47379a538b207138d610 : SUCCESS in 4m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d42e8b81fe4349e8adf9aa9f98052006 : SUCCESS in 10m 52s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8c22ac5d3aa5436f8a2df282a77cd5f2 : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7dbbf4945c2b459f9cf8659a946ac7cb : SUCCESS in 5m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/de11cd6f35664a12909cab5cd8275340 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d84281aba1124fa9a55491cee8d52c9b : SUCCESS in 17m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/000f98e5259a47028ff83bb5f36b5aad : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f1c6d0b816b3496ab751bfac58521e01 : SUCCESS in 9m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5e9a599b1b2e452b9168c8963c602cd6 : SUCCESS in 13m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9ff90182dddd40e68fa7de0052e3c912 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/72d9945559314f25b87964aa20488d47 : SUCCESS in 5m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1cc9b4f09c0a4c3bb9ad906503faa250 : SUCCESS in 4m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/640a87a8e8a342888080f0e6ef99f162 : SUCCESS in 5m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6fc63239ec154209b168fa16c669dfff : SUCCESS in 10m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7e891e6f697e4076a28ebdea614ba7d2 : SUCCESS in 33m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a0f9d780c947484eb175e0a6b157b163 : SUCCESS in 7m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a75af28dd4f84e8ca55cc45124c92693 : SUCCESS in 19m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5d9246b4402d464eb3f060b80171aeef : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79dc98877d13459ba67880f54b8d01e7 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a88f86611b34d64b7f1b8c71f54bfa8 : SUCCESS in 26m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0724c2c59fb34c13870a5660b18612a0 : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f1fe0c1326fb4ab684987f0e14b7b06a : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a192175cc66428fa9300d3531482d9f : SUCCESS in 7m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f2d080a355fd41ea88f6edb17adf2f6d : SUCCESS in 18m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/861fd97a606c4ae188416af32dc0bee3 : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55b44350a4e24ef4aed7854008eb11c6 : SUCCESS in 37m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d016329b66754521a6bdb1f53a5ef2cd : FAILURE in 45m 51s","accounts_in_message":[],"_revision_number":2},{"id":"83bafe0e5ff51b766ca26e63080d34bdb4068721","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-20 07:17:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a71b8258d948201916ea4aeb7f4cb9edc0932317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 08:39:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cc0b8df921d41b0940fb51b943c7e4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e6446bd2744abd9d53690e12a95015 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2638aa1bb4ef40d599c6cf35321543d0 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afa5be5a0e424ba4903fb102d064dac2 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/304301623bc14a9d8dbf5b085c369f86 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e802c5b14a0640ed9ed562120d80728e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08a334507ba1488caea4dc9e2efdf268 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0f5746f882a4f93a43350b331a55280 : SUCCESS in 2m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b601656b40f04c13a8df33254c46abe9 : SUCCESS in 3m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/20d2c43c06c84a739c810173979a6d8f : SUCCESS in 6m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/307019915dea463bb846f1a6557444ce : SUCCESS in 3m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3426bb9d098f40a3a0fe11c333bf6228 : SUCCESS in 3m 32s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8c584f64bc72449c823ecad02529f252 : SUCCESS in 4m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ce736669d0c040dd9dc32ac45250ab0f : SUCCESS in 6m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7f04b3febc2a4ef6b754962dcff353d7 : SUCCESS in 5m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8d391b62361f4fe68a2f5ae47cded2dd : SUCCESS in 7m 05s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b36d32e42d604e9e8cac85a27b141d3c : SUCCESS in 3m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a6991236e3bc4bf3a1222e7b9c73969d : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/07b480fb023747a78665ffa63d391509 : SUCCESS in 3m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2558f75e48344f9994ce197e8a37305d : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a09e03d2b723411a85ae7362690eafca : SUCCESS in 6m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2a2b6e3eac2243bc90916cb6964279fa : SUCCESS in 7m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/37a9f7ab23b340b8a5c153c393d194e1 : SUCCESS in 11m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d67298e69d964456bc407c4c65adc75f : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8e909f9ad57b4217a7ca0ae72345c33e : SUCCESS in 11m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/acb775d965b54b63bbb39f2d881c9e21 : SUCCESS in 3m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bc114aa77e8f4a6f8408ed575a80e86f : SUCCESS in 6m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c0cb48ab7f3c41fd857cd575c1d19c1a : SUCCESS in 11m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3f27a47b27bf4a8ca01c8e9664922469 : SUCCESS in 28m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7ef77219263e4ca0bd3d47c23f9451c7 : SUCCESS in 9m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/685445353f9546a5bc838bfe8c5e4aba : SUCCESS in 9m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e95478a20cb84a4b8e496989cca2190c : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8702b11a24f544cd930d9d1d3ffde5a9 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7296201a3244ba5a5742eb176abdea8 : SUCCESS in 26m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5d6d1de108114d92bde8106a4951b19d : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a6f963e228bc4c11b06d1f611da9ea47 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1025d0caa4454516bee66b3ac1433c2b : SUCCESS in 8m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bffcb95d2e6142d7bd2897fab0aeead0 : SUCCESS in 19m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6f2c5653b82f4ec79d898b5a044a8d57 : SUCCESS in 7m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2edc88b412104e96919da75aa27622fc : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/042f257d96264c509ff568ce74e1bf5a : FAILURE in 1h 16m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5f2af963e6b8714e2637c2e3f746d9cf78a73480","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-21 20:59:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfd73e61217931c4018f2a9585d1be259159c6bf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-23 11:45:09.000000000","message":"Topic set to bump-testtools","accounts_in_message":[],"_revision_number":2},{"id":"eeb3b7d68a6f43873367e48aa2cfb4461d3da09a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-23 11:48:04.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"403bd27b732f1b1970a34033734a1bc8608d0c2c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-23 11:48:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c7ef19d9fc8a70565d1757a4f422f520cc75942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 12:56:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2832d6ab519b4e76b9385b33a4ee0328\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5588cfeaeba549019991671bea5b9a47 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e7e9ba67d94e3087328655d7428d4c : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8daae65fa2f94cc29ff2b4794ff0d2b7 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab9e6f903029488994a596672fcc0b78 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1b96211399e4fb5abf1070ed988a265 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2385c037d994a058bf1fa8de00c19f3 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3f6f05389a64fa9a7686b246d7d584d : SUCCESS in 2m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fb7c031ec96d4b82aa3f040d5661bea4 : SUCCESS in 2m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84516ac9c13b48f486b220e6d218af87 : SUCCESS in 3m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d29260213eea4e4b9895320a4f055979 : SUCCESS in 5m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/98d04672eef84ea69e1479580676de07 : SUCCESS in 2m 54s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/acd4ecb31f894df19d7c74e5fb4b9853 : SUCCESS in 4m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/487361b31f01402f8e38382422b2e207 : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c61c4f2786fa473dab6085ba59114b4f : SUCCESS in 3m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/65ffb6f5cecc48fda27a4584792f088c : SUCCESS in 7m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8317c16b699b4d34bc8d5e97101f77e3 : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/de3baf8a34cd4988b321f08b49d94d1c : SUCCESS in 6m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d5598d6849634ff8b3d8234c5f8bc447 : SUCCESS in 4m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/25d07e7fdc0347fa80ba5f2b8290a752 : SUCCESS in 17m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/237f8def36c14dbebf4ed0e117f465c8 : SUCCESS in 6m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6b747e9963c14cf28adeee601f074f6e : SUCCESS in 8m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a4d54993556a412689a1b8ec596d3964 : SUCCESS in 8m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/283c8be96f2d48c2819c0727624ecc68 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/babbe5a7e53c4739a8cb97b7961c3c4d : SUCCESS in 9m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e302f6a180714873a7da578dd43bd2e1 : SUCCESS in 2m 35s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/79a07275cc12453eb1ce0715286682d1 : SUCCESS in 6m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/905b9a5d8cc34681bb88afbdb7905be6 : SUCCESS in 11m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c72c72969e104be0a4f9c2911229aad5 : SUCCESS in 23m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2416bcf97e7245d1875ec36bb935462f : SUCCESS in 7m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/44fa77df466c4ba794ca41524e6195c8 : SUCCESS in 15m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f07045c56e4440b38f78f8504aea106a : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62c3cc7c86684653b7aca4cdfb328944 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ad559a1e8644daeb98686a33a79592f : SUCCESS in 21m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4beafa77974e4db6bbea557fd0e25153 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4c6858a6a9fd4badbdc9dcd0f489d48e : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4450be92e59947518047f028871c8f08 : SUCCESS in 7m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/133bb23076bb4f1a95eae8615d2dbb61 : SUCCESS in 12m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bb2f6e37701243c8bb02725b96b6fe16 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3583eb1b4e754e5585fed1e3921ba808 : SUCCESS in 1h 06m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71dfbe3394ed4240ad21f75ab2b7d7e0 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":3},{"id":"8bba7c02770f05c225458a9ab158aa4efad63a58","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-23 13:02:10.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fca271f5902075463570e9cc466b4cc912fe7252","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-23 14:35:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd4cea870a47d9ff6b3db9b3b22d9cf59b5ee618","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-23 18:24:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"372374090295af0f4cc0086109138f9af0bad67c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:27:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"358438f9b7a795cf4d7e4efb814b7a9285e3e77b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 21:23:56.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/977616 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"656bb805c7d9245bdbb07a1fa63d977316c8832c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-25 07:25:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1bbe92292ec7161920931034ff2c8eea802a967e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:19:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f85610b89bb485a962e12e0e321e8e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b7fde962264c34a320490108f6ea19 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afaaf4b6ec1242cd892639f218eeb6d0 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b30b8964e6e41568565da5ff546055b : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0e5a0735d874968aeff6b333878e7fa : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b6675bda5764321bc309e70b795df37 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c9b8ac30a2c4b01bb10571d57b0d806 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c533d6853794859ab9ba1a5ae4ef2fb : SUCCESS in 2m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b8785a7364074c5ab032c9d199c7761c : SUCCESS in 3m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14f0a405c1ee45f8b3601950cde951f2 : SUCCESS in 5m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4dea654e62354cdea8a54848ac5f183b : SUCCESS in 5m 56s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4d7af91c18ea4f77b2ab56ec77f66485 : SUCCESS in 4m 58s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7a625a5ddd4d4bc499c9bc693ef95220 : SUCCESS in 2m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fea733507e16428580f94aadf048b202 : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/42da6a7eba2c4e6ab06b407dd788951d : SUCCESS in 3m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7d711784d3cb49c78cab3316093516d5 : SUCCESS in 12m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3ebc9446718c4acebd0962d131706081 : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/61a69a455e0c4496be90901673d4d949 : SUCCESS in 5m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/71096f37f5524e17850a3eed82e6e72e : SUCCESS in 6m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/090de27a768940b9ad62ba7e09dce21e : SUCCESS in 17m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c9b006916e8491f84141f264b298e61 : SUCCESS in 5m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3558c6a13a404561b2503dd146df893c : SUCCESS in 9m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/84e179849d154cfaac77db45cce9c34d : SUCCESS in 11m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d2bc5e188084464f82416b3e944719b0 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0b7f948fd17a4dbd88c92e7090640094 : SUCCESS in 10m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1ef08a9062254741bbfce79f8236d974 : SUCCESS in 5m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d6461a63491d49cbafbf034e9284bc38 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b312a5faf9af4d9fb060925b9042ef46 : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/703cc0e368ba4adea8e503c4f4a37915 : SUCCESS in 27m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3c886cec56104d3fad569f7f9f0168cb : SUCCESS in 3m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/80fc13089e8644edae0a125e92cae763 : SUCCESS in 15m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e6d5b0023d864e5b92208820c6bdb6fa : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e63721005001407aa670d6f99b7c9ed3 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29a7c2e7b6c747858d9ec6d9b62be0ad : SUCCESS in 26m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ba79004798a445e7bf676b02bdcf0035 : SUCCESS in 3m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ac753786df794e1887d476ac5d6a3d50 : SUCCESS in 2m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89b024b70f7e4f6d9e4f73231f230767 : SUCCESS in 3m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4ad0b64f27ff47a08e1b5e217f6d2747 : SUCCESS in 13m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f21e984336cd4b3592d0506d21adfad3 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9766d18778d04ce8b55f713c09ff4bcf : SUCCESS in 51m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/163f3848f8c3403eb4724dd7000e8584 : SUCCESS in 1h 47m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0d5687d388331aceb368ef94040fc0ee94b51589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:20:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"330a930c16de784df5de3fe88115826b44910db4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-25 10:58:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98acef921614ff3a6d56b8ad8b902e04c08d129b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:25:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f2af5067ce49648cee9303a443fe6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790222abf0d242448aded15306ba054c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b280f4c93674f7c96f29d638627b232 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d6e13c28090488aa67fedfe29de9310 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12abfe790cc94c08949cdc47ea64d12c : SUCCESS in 2m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80e500c4896d4876b905489d6fee3485 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb02d977e29c4de88afdd9a030c32b52 : SUCCESS in 2m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1aac558da94247b0927a00044586424e : SUCCESS in 2m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9e2619ff50b04ff08d2af54e89446052 : SUCCESS in 3m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb589ea92b774fb8bf6543804274763f : SUCCESS in 8m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cccd458a08bd4590b030988de3e0e323 : SUCCESS in 3m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4578b7894e1248efa14ff9f3d2944ff0 : SUCCESS in 10m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5abbb435356d4fcba746789a4e395a92 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/061de8b02916411ca6fc1057c70b3058 : SUCCESS in 4m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/39df4d4e24b44e599b89a79a06280aac : SUCCESS in 6m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/331edd2e47b6424fa72f6550e95f3fc9 : SUCCESS in 17m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ab4d9c586d64025be29ae2858d40cd7 : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/64dbb7454dfb4e5bbcca0fabb56c8fbf : SUCCESS in 9m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c7806cc0464148968d94098aa8d0de40 : SUCCESS in 7m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/32fa76dd12d24385b0b3ce7e75fd20a4 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3f94e56573df48e4a5393a12a439ec58 : SUCCESS in 11m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/940eb56a44224fab9226e84fc34e2bc1 : SUCCESS in 29m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/862332b113a24b03bf90543576844efe : SUCCESS in 8m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0081aea5f53d49a6b75c78e456776c5b : SUCCESS in 13m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c8b0139c009d4b5abfb98d7e2c679e35 : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/775e95ce3c5d4ab5b6dbead49b766f97 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81503e501a4143dc93633f73da5cd63e : SUCCESS in 29m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/136362f95f9944ea93ab88ae3e51dab1 : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4876d9d02ddc45bcb1226577ea4f9509 : SUCCESS in 3m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f309a998d96940a998f0ae1f29e1f0ce : SUCCESS in 6m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/81c89ad96f624eecaa2a5190c603c8fc : SUCCESS in 13m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a0d066e06b5a4583ba281bb9611a4bb8 : SUCCESS in 5m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e105b825a8404297b4ef5d4825ff04 : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/275b2ab49a15444c9c299f1dc53826b9 : SUCCESS in 1h 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e28cdb67a32f8644a6cc703ba0d5f2d78a47e5bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:25:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"529e5ae5d9f278499be09d610ed0763ce355b716","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:26:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7bf9aea66ae4917930dee9c5178fc5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad32c34915844dc4adf9b0cfa350e896 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3},{"id":"6ea08102e1a0d174a8a9018b77302a966b916597","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-25 14:33:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"184a56937c4d6fab132ee08e2fd44ce641993b38","revisions":{"8b2fc9047d8a6b808828119956128cea06ac3dde":{"kind":"REWORK","_number":1,"created":"2026-02-17 13:31:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/977082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/977082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/977082/1"}}},"commit":{"parents":[{"commit":"38810fb815a60afeccf61b2b5a04449fe948dc82","subject":"Merge \"Drop Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38810fb815a60afeccf61b2b5a04449fe948dc82"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-17 13:31:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-17 13:31:11.000000000","tz":0},"subject":"Update testtools","message":"Update testtools\n\nPull in type hints and misc other fixes.\n\nChange-Id: I263137e527ff83d45e7068e02e5bf2f99bc01591\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b2fc9047d8a6b808828119956128cea06ac3dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b2fc9047d8a6b808828119956128cea06ac3dde"}]},"branch":"refs/heads/master"},"49ef354d01385aa6c11c5d775059da92910a6ad9":{"kind":"REWORK","_number":2,"created":"2026-02-19 13:06:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/977082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/977082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/977082/2"}}},"commit":{"parents":[{"commit":"38810fb815a60afeccf61b2b5a04449fe948dc82","subject":"Merge \"Drop Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38810fb815a60afeccf61b2b5a04449fe948dc82"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-17 13:31:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-19 13:06:28.000000000","tz":0},"subject":"Update testtools","message":"Update testtools\n\nPull in type hints and misc other fixes.\n\nChange-Id: I263137e527ff83d45e7068e02e5bf2f99bc01591\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49ef354d01385aa6c11c5d775059da92910a6ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49ef354d01385aa6c11c5d775059da92910a6ad9"}]},"branch":"refs/heads/master"},"184a56937c4d6fab132ee08e2fd44ce641993b38":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-23 11:48:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/977082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/977082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/977082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/977082/3"}}},"commit":{"parents":[{"commit":"38810fb815a60afeccf61b2b5a04449fe948dc82","subject":"Merge \"Drop Monasca\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38810fb815a60afeccf61b2b5a04449fe948dc82"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-17 13:31:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-02-23 11:48:04.000000000","tz":0},"subject":"Update testtools","message":"Update testtools\n\nPull in type hints and misc other fixes.\n\nChange-Id: I263137e527ff83d45e7068e02e5bf2f99bc01591\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/977616\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/184a56937c4d6fab132ee08e2fd44ce641993b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/184a56937c4d6fab132ee08e2fd44ce641993b38"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
