)]}'
{"id":"openstack%2Frequirements~977136","triplet_id":"openstack%2Frequirements~master~I3070bc488e37a378f938fc943efcf7c6dad7f4c7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3070bc488e37a378f938fc943efcf7c6dad7f4c7","subject":"update constraint for oslo.privsep to new release 3.10.0","status":"MERGED","created":"2026-02-17 16:18:34.000000000","updated":"2026-02-18 13:13:15.000000000","submitted":"2026-02-18 13:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977136-new-release","meta_rev_id":"64b5e3e0ce2238b4242a82ca910eda0d49fe7b52","_number":977136,"virtual_id_number":977136,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 13:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-18 11:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-18 11:17:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 18:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 11:17:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e120e206bd4ca8afa7aaf83812f9503a74e6468d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-17 16:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7e93566d95a02bb32c42d8ba0d0ef7119052e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:08:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369b21d90e094034a4dc90c0b1730c9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78404eda76a5480b892ed5768ca42678 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd1cdd344a94a13ba73195933c1f79c : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12d099c83160465bbc1ec9997eb8d9c2 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47e8b82247d84f4da1b710df1c2087d9 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93c34de0903b4a8f88e9b591e0dde708 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2e846bbceab46848be37d374632b86a : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0b3406d3645e4cf7a0e13737602d30b5 : SUCCESS in 4m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a60fc35d58e841908f0a9790823f693c : SUCCESS in 6m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3a7e819b0d4449d9b941984e27133602 : SUCCESS in 6m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6a276fd366764e389be7ed3f1d498502 : SUCCESS in 5m 26s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3d7caacd16fb48498e80a10d51c030b0 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9432b8dc420f47b78bc35fbde34cfcab : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/88ec103d324c465db651d56a69f681e9 : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/950ef5a279604a8f8cbadae6fe6957c6 : FAILURE in 11m 41s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1697e1a9e3db4209ba4459b24033e521 : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/712d10f554054d62bd5a09532ecf92e9 : SUCCESS in 5m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/05bd1b1a939c4e778a3509c65d944ad8 : SUCCESS in 8m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/db97f2cb90804edca9658fa7f32ad6cc : SUCCESS in 17m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8f01bc056c44b49be772c7b5eb92e20 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7adbacc2c8274e37b1aa62d862194d7e : SUCCESS in 10m 51s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d687ddc82ea44908b3a69f3d4cd290fb : SUCCESS in 13m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3bfd9f1a24884a6a91ff45f282abe337 : SUCCESS in 3m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8bf2c6c87ac541b4ab84dca5427f0d30 : SUCCESS in 10m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/574f8e4bd22b4f2991082f589b1d502d : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bbbad470513c4de198d34d3f805a389b : SUCCESS in 5m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/70bd7841d8e24cf78889203972a8eb77 : SUCCESS in 10m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/61ac8c72a6d04e01b05299a89828e34c : SUCCESS in 26m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5f73f78cc35487daf70a87d0a46cb4a : SUCCESS in 8m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9f4f632f05e94c9fa4353ac54a8d9197 : SUCCESS in 14m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/af3cd59f34084198b161df27bfb7359b : SUCCESS in 3m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6223ade889804d99a8228f42275bd645 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75b7bf95a91c4f378841ec061271364e : SUCCESS in 28m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/feb4a9e51d6343dfb56c615b2d7df88e : SUCCESS in 9m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/065f4225954d4dc9bc9e2bfc51721597 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bbd5f89676e4408ada831ec98573907 : SUCCESS in 7m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e062ffb08ea246b4b5ddc39c6949d894 : SUCCESS in 6m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4066b8b6b8a6441e939f9731b31730c1 : SUCCESS in 5m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d074da67c5034268b200ee315917cdb8 : SUCCESS in 36m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac98d4b3034b4f8ea6c694a284938f1f : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0660f63d9d4aca8580baab5a2f767790b4f8edce","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-18 11:17:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"983ebbbd686f2cf41b79dc89b9819f8f4874ec17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 11:18:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e122ba8b0da630928c24dbab57281719f050460d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:13:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6a489778728489284829e3b9380dd2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba147b940f14a8c8fe050a6d160d80f : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2d23ce17d04e1da9ce929c88311d55 : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ba52b0077af4c1097cb87076efde952 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a4729fc0e5f4cfbacf412db05dcce00 : SUCCESS in 2m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e5bc62d75ae4bfba7b6bb7957915c25 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e2f0d970e4e47a4b567c01873bb0998 : SUCCESS in 3m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e9f85e4a691a423ab315807a2b350cfd : SUCCESS in 6m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/26a770a9949447b09f7d8f66ef4bbf55 : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/96d9d2304e0d4fa7b2d2551c3e95eeec : SUCCESS in 6m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/07bc296d813e4151ac86e23a64e79947 : SUCCESS in 6m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/356660260a684d7c9ea2550a8702300f : SUCCESS in 3m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c3256abebbbd428aae03dc42274a70cc : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/70820817c4af4b3ab4ee049565dd7073 : SUCCESS in 9m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0160964724154a8b8c12f5882f9dae95 : SUCCESS in 16m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a44aa6f7ce6411baf183e7d364365c7 : SUCCESS in 6m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b48acd99b4be43e99bbf489036dacd6f : SUCCESS in 10m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d2265d1fde0549b8a6e7068755676ca3 : SUCCESS in 8m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f5256276ddd748098fefbeedf761c129 : SUCCESS in 2m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/73d51c0a1d084f2da1c47daa20c59afc : SUCCESS in 5m 45s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7741a3f160ec4a778e1b6e31580a0311 : SUCCESS in 22m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e61e5d3348048c98fef09bfb6f0e00e : SUCCESS in 5m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d35c461b92ba4d52b21e9ea8814fea6c : SUCCESS in 14m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/68e6fda706a641b9bbfb7ed9b579e7b4 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9342d1acbcba4a0db4a516477db3586f : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4870e8b9e8c4816923c125c03e1d7e4 : SUCCESS in 29m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/91612c713a9146879b11a6fe46ad378b : SUCCESS in 7m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/20030a55eaa24acb8b0d99fc51a31dfa : SUCCESS in 3m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f095231fd604f61a846ef1d19a1660f : SUCCESS in 7m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/71e108cb8cbf4e0985bd8153414e5946 : SUCCESS in 19m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0847b4107f074b4c886d7303b1361390 : SUCCESS in 3m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2d4fc0d92fa463ea7dda9cdb5ef7312 : SUCCESS in 1h 05m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f47ffa0bfad4652abb90629f69e6e6a : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"64b5e3e0ce2238b4242a82ca910eda0d49fe7b52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:13:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33a238e86ef7ba48db25d511baea7b204cbc7bd1","revisions":{"33a238e86ef7ba48db25d511baea7b204cbc7bd1":{"kind":"REWORK","_number":1,"created":"2026-02-17 16:18:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/977136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/977136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/977136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/977136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/977136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/977136/1"}}},"commit":{"parents":[{"commit":"8b5138c1baee3589e25b9d63fede298d2b370dc5","subject":"Merge \"Make requirements-tox-py313-check-uc job voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b5138c1baee3589e25b9d63fede298d2b370dc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-17 16:08:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-17 16:08:33.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.10.0","message":"update constraint for oslo.privsep to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I416ba4778c3d490b0c8455f6acf22ca4df186f96\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3070bc488e37a378f938fc943efcf7c6dad7f4c7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a238e86ef7ba48db25d511baea7b204cbc7bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a238e86ef7ba48db25d511baea7b204cbc7bd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
