)]}'
{"id":"openstack%2Frequirements~977137","triplet_id":"openstack%2Frequirements~master~If06fda16383c43b3482fd3ab32b1ac9491ed1630","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If06fda16383c43b3482fd3ab32b1ac9491ed1630","subject":"update constraint for oslo.limit to new release 2.10.0","status":"MERGED","created":"2026-02-17 16:18:46.000000000","updated":"2026-02-18 13:14:17.000000000","submitted":"2026-02-18 13:14:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977137-new-release","meta_rev_id":"9a128226c7a9dcbf82728b8947579af6a5a88b7d","_number":977137,"virtual_id_number":977137,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 13:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-18 11:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-18 11:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 18:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-18 11:18:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"226b00250f82c9fafbe4b0db16283a684168ee56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-17 16:18:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4b1b3321257e4207e2f7096406339d299a20b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:38:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbdabd069ca426884b7f4d2ff5fdb6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052cffdaa48d4894925895b50910b2de : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff6c95848374fe582971d889f903c85 : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7622b2fdb50340f1a6d68d5d12f341c0 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12718cd1e62f45a49c3569d50fb4e17d : SUCCESS in 2m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/962a27338ce54c2c88a97bbd9931feb0 : SUCCESS in 2m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a1372f9709647398a7cb8991b615d64 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fd1529df0623477bb627fec401886432 : SUCCESS in 7m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dbb85754946441368dc7f74781dad9b9 : SUCCESS in 9m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f5d09fbe8808483dac5140f51867f490 : SUCCESS in 8m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/de6b88378b514985807ab2886434523b : SUCCESS in 7m 09s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/dcdc3f815f3a4d3cacacc77ab2532aac : SUCCESS in 5m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71a86cef0a1749829fda1bff663b17b6 : SUCCESS in 3m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/aaea35da3d124ad0b16e141efe2bda49 : SUCCESS in 7m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/79734cdb6b884fd5a108530c0e1cda8b : SUCCESS in 10m 50s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ed36f877e8624bf1b092d89352ebffb6 : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/22e1c0c9ec5e4508a3914066131bdd0f : SUCCESS in 5m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ea2558df74de4e1e9a2c988f42fc00a6 : SUCCESS in 3m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/22a536d2e0b44951b3d110567c50dae3 : SUCCESS in 13m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fdbaddc1fc64f599fe76c6bc7f4527c : SUCCESS in 6m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2bb7a16833994e5998b8b347b3009588 : SUCCESS in 9m 09s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/55ccbd132a734f05bf272448bc4b474c : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1b7ae05a4248471eafcf9334f54cb0f4 : SUCCESS in 2m 38s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cea8881a129e44a491d496a77d1b20d4 : SUCCESS in 9m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/990cadc96db24bfb97d96f9a2421f3b6 : SUCCESS in 5m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d510f59d2da0493a852061ea147e57dc : SUCCESS in 5m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/570e32e615584392a26df62a9e2c1218 : SUCCESS in 11m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ab6945c71bc1482f9344b0a6f550cfe4 : SUCCESS in 27m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e23cdaaa46c14c1cb1dc1dba8faea01a : SUCCESS in 4m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/aa442ae7964a4ed38c87f8d1006df7de : SUCCESS in 10m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d8abde4ef01b4e22b1ea1b7f88d459c4 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/558c5b4510e64a11858e951fd83ccb41 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3df3f61e38b48ea8c3316c854828f55 : SUCCESS in 22m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/776254f0eac04c6d80de738366f9bf2c : SUCCESS in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5c951f4ea00a4cf0852c2481d1151401 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88d5db500e2b459799b9a4672e4fb3c7 : SUCCESS in 6m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/310a135fa7a44fedb7636ff9124d09b4 : SUCCESS in 17m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c2b9c33c69b04e249604c01fff9cbb0d : SUCCESS in 5m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5110b17c8e8744b69576fdc4432c5bbf : SUCCESS in 33m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5323c657e7a04474b501c3a669dd575f : SUCCESS in 1h 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7c1f92668d0b29ab083d318a06610d0de063427b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-18 11:18:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7642ce80ff496bafb79827622185e439cd19564f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 11:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d888fa62ea8fcbb34b4da4513909ffca8775124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:14:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d46897c799547ca97919d98d9846f22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71b273e2ebf4835aac1144bcfd6a60d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8858b8a327d47ccb3e8bf14b058a65e : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0fbbaedc0b744d1b1d9b5b0aca30a61 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49d4d67283eb437c8b17f463dba99c60 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b48a4a29d0f4896a3bbc12660e0ff51 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b3a4a70d2474fdbad3baedaec53f195 : SUCCESS in 2m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9786c1e08fd41f88282a4e0584e8780 : SUCCESS in 9m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/02de883e63274466972edc02323e9827 : SUCCESS in 6m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9cb6001d2a0143fead56aa64f13b3bd9 : SUCCESS in 6m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b4f0f2dac0bf40a5a761776e61304ac3 : SUCCESS in 5m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/36299e23b0e44ff483f823338e6e701e : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fa5b6de774bb4f27a31ec41768baa38e : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9f86c0d66741497cb0d42304942cb8e8 : SUCCESS in 7m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a3c1dba7fcef414ea498d576a4f6ee4e : SUCCESS in 17m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a19a229af804491aba77bac0a1539aee : SUCCESS in 6m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d9fc64a1ae944aa95f6612d7b1cc0e9 : SUCCESS in 8m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/94f88a10b10a4ba9bd6f430c19b3597d : SUCCESS in 11m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0abf71cf4f364b3bb4d3c99baf8746be : SUCCESS in 4m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/099aa9bbb62b45ce820a460b8271bf4f : SUCCESS in 5m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3150c53bcd9c4cf391722dd2666fbe19 : SUCCESS in 23m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d7b85ddb5c643c891972fcee060958a : SUCCESS in 8m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8b55aee812334b6cbe501100034abcb4 : SUCCESS in 14m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b01a7948c4e043cfa61dc65407d3a568 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2670fc8688a41f786a7befcce9d71c6 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ae4a948df484a2593147953dbae2979 : SUCCESS in 25m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/adc6b24e62e948a8afbe84d2b0b525c4 : SUCCESS in 7m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4a6130d3c057420b9d45d496eee00941 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d648dd286a144c48aa35f0d7eaf2540c : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/748eb4f55d24419ea288caffab4e8a10 : SUCCESS in 15m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bfb7a1a396c94ef1a31599d2a009c8a9 : SUCCESS in 3m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba9d731a5d0942428cf7cb6adadbb9d2 : SUCCESS in 37m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/831ba37396424f22ace4cbeaff0ac9bd : SUCCESS in 1h 50m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9a128226c7a9dcbf82728b8947579af6a5a88b7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:14:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bd023978962745667c446e84e0ba91c29839aea","revisions":{"4bd023978962745667c446e84e0ba91c29839aea":{"kind":"REWORK","_number":1,"created":"2026-02-17 16:18:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/977137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/977137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/977137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/977137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/977137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/977137/1"}}},"commit":{"parents":[{"commit":"8b5138c1baee3589e25b9d63fede298d2b370dc5","subject":"Merge \"Make requirements-tox-py313-check-uc job voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b5138c1baee3589e25b9d63fede298d2b370dc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-17 16:08:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-17 16:08:45.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.10.0","message":"update constraint for oslo.limit to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I82088a4d7117846bd1aaa535724826a884396e4b\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If06fda16383c43b3482fd3ab32b1ac9491ed1630\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bd023978962745667c446e84e0ba91c29839aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bd023978962745667c446e84e0ba91c29839aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
