)]}'
{"id":"openstack%2Frequirements~977302","triplet_id":"openstack%2Frequirements~master~Ib55269dfb476eca64936bd7e4f60ac9d3e8b4c40","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib55269dfb476eca64936bd7e4f60ac9d3e8b4c40","subject":"update constraint for tooz to new release 8.1.0","status":"MERGED","created":"2026-02-19 09:55:37.000000000","updated":"2026-02-20 11:53:06.000000000","submitted":"2026-02-20 11:52:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977302-new-release","meta_rev_id":"bfd12bdff1e2dbfd1148ed225d3c7d51ae61d23a","_number":977302,"virtual_id_number":977302,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-20 11:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-20 10:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-20 10:07:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 11:02:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 23:18:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-20 10:07:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c7f3fc35e34eda21d7bb1ca060491b860afc4aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-19 09:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf214f60c9da6e78d10306ee8b1420090af19030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 11:02:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6516f72c71a048659eb3697a1f07f1e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715d1916ef494239a08465d7b3b6656d : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530bbc947cb34888944d68901078c768 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a07163304d1240128ceb2b9ad046d319 : POST_FAILURE in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8c9c03409cb48a5823f0f525373092e : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff7d9dbe673e4ea69fdf1bbaa4919825 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4e0ce46e8644050b401acf3777422f8 : SUCCESS in 2m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5b74acc28f2e41d1873a6a06fa50242d : SUCCESS in 4m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3fe9aaa28454dcb929ad23f7cb693e6 : SUCCESS in 5m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6cfaabc9583d496e9fa9b49dae935926 : SUCCESS in 6m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/85656184a30746f3bf1f3a0258a31f01 : SUCCESS in 5m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/445f2cf418144112a4e9d395f06bc03a : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9181a75939224611987b6df0c4ea8a6b : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4d6a1074a1fa40328e2faaa5d99a0540 : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/987e77374d6144f5a437d4ee87a2a5cc : SUCCESS in 11m 11s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/14bfa973c61a46b09610849437890fb2 : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4476b4907d8e41fea126b760d1cab756 : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/427a11eb7f44408480412fb5e9e55826 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d6de404cb41a47e4b58fd416ba2becf4 : SUCCESS in 18m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/733bcaf8d86d48d48bbb64fa24312fe1 : SUCCESS in 7m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/32c9cf4b044a4ff78b955184f2a0c120 : SUCCESS in 9m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f61a030230ad4c4896404dc7e2ebeab1 : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0074341298004ceda0d1074f23317f5e : SUCCESS in 2m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f1820bb042a341e98134b33ed87a65b7 : SUCCESS in 10m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/296ffb36caf04ec991781e8f98b68230 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/70d1c6eca3e141aa958ff38705b81318 : SUCCESS in 6m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6d3cf47ee14646d39729889073a1e06a : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4e5cadf636cf457eb06e3a3a292f19b3 : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/580a28124b7b44c7a7c498f758b1fd4a : SUCCESS in 4m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b89d4187fc834783b579805d8ec27677 : SUCCESS in 13m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f5136fc970d044fa816ab7d8c9b1f1ed : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39e9d867d9e64a68a6b57235d5e172c7 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c76a20d4c7794ceea646807e7b65d15f : SUCCESS in 28m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/73ef5e46c9424e5d91c3061c914de2ca : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4a9bb40b08164aa785fdd2b3944fd348 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bfc5fba5675454ea9c1fe17193b4307 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3e9a9906926b44fbb1242aa848f2e291 : SUCCESS in 19m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c291354bb27148f49b121965dd435c10 : SUCCESS in 7m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eba4fa4bde634ea59e9957b5a82b05c7 : SUCCESS in 1h 01m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09f60bed1f7446ad92a3c3fd4c617b69 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b0466d567954e0ea1bd319858fe3c05e284b8e93","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-19 23:18:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c67a4de692dedc3e18aad39ff2442014951f5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5138573a4dad48449ce8cbf9ca013384\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f466466296094486a9a31682e7ace4de : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8894fa8ee6884998b55fe12307a6bbb0 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a913866544e34f5492024f3dfde55fbc : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac68492e5fc24839b7f7c9e050c532aa : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/903a0dd24dde4054b0b9ddbae7a8ed8d : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d43cac6eaa5b429ebeed4742f2f9ea60 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd61f1c1fd9b466186bbdab757cca478 : SUCCESS in 2m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/50a4c8dbb4e04a188d69a225a7a3c9a4 : SUCCESS in 3m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3e0f6f680d3c4ac292a3f7997f708b28 : SUCCESS in 10m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/68d2315c404348d79e4dcf9b355b3fb2 : SUCCESS in 3m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8c1cea5a755247f9b0856d06357740ee : SUCCESS in 5m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1b754a889acd43e382d76d94b01adddd : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a0a9bf9efc584a718facf853c9943fd9 : SUCCESS in 4m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/aac223740147468a8cda86ad889329b0 : SUCCESS in 2m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ca55a5f8d8ce49b2848b197af65bd710 : SUCCESS in 6m 44s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5211e3759d2d4ab39ea532743e6d6f43 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/904885ace2f0486db01d7bc484de6398 : SUCCESS in 5m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/868a89609e074266af01b75110b1cff0 : SUCCESS in 3m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0693f1c7db7f4899a7e49e6391b4ed6b : SUCCESS in 16m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da1a6903086042829f5c49342b79c2a7 : SUCCESS in 4m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4e9edd40cc4040f1a7d05165bb1f9a5d : SUCCESS in 8m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ee7583848e51426d82b071dc926c09af : SUCCESS in 12m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3c30634df2eb46888d06730443ea86cf : SUCCESS in 3m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1c4fb417489440f4ada58bc97b4025ad : SUCCESS in 9m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/15426362ccf946bf9d2cfb2d58c8de9e : SUCCESS in 3m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3052c64896b54c71b28ce7242be07b19 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/21a7be637b7144aabebc0edfbccecafd : SUCCESS in 9m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/81a50ed3cf4c493ca525a4a5411a0d40 : SUCCESS in 18m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/965509fc378e412687c02c946ed25621 : SUCCESS in 6m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6bb0f8c696054a1a8a7a7ab8f5fbac5a : SUCCESS in 15m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a5223aa9009948858e23aab947ead6c5 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ffc875616e24554a5879151845b59a9 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c74eb8cbfcf94c8b9f7a3cd658df26ac : SUCCESS in 28m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5fcff263b7b34fcc8f576cb4521cb9c6 : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0da71a3c4d8d4842a250a80e1e10a3c4 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f03c74f676c4847bf8af06938dee9d8 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/002c5ef06e204f3bbeb0a9c7eee3d33f : SUCCESS in 19m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a8d4fe88c62e43a9b2e3ad2f48d075d7 : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d9fde8be8ff40e18d9768151fcaee29 : SUCCESS in 1h 09m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5de9772433742ccb6bac874e358ec76 : SUCCESS in 1h 52m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f8df33f957e8fd699838305cdbc215b51e1fca47","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-20 10:07:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7912446e490eab0c3947d9b258047bac44b1bbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:08:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81fabc6490821b417d5da7a6aa57ba308f95f8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:52:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd757037178c4f668b106a848532273f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a205cfe9664519a101b3987e8026ca : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3fd9b9597ad4e9e96e58430653c7895 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cc5fff85d12497e94938396fa75dee1 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89953e24a73444ef823c378d311567fe : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e85e3c56a9ce4f92af845d34032a9fab : SUCCESS in 2m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abde509432a944e69e74ef24adc774a7 : SUCCESS in 4m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0211fe55d4b4624becd1ef16195cf6a : SUCCESS in 3m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e151acdb6239444bb4914f46fb10c88e : SUCCESS in 5m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/30e6404fadfa41b581631538158a888a : SUCCESS in 6m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/beea449900a34db18053f8bcc146fd1c : SUCCESS in 5m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/146a963e4caf40c287701678619e1951 : SUCCESS in 4m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dffeb6e3b46945a290168b5009dfaa11 : SUCCESS in 5m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/707eaca7512d4b499a20229354c3f70e : SUCCESS in 4m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/375736e8f6bd4eedbe16634adb776859 : SUCCESS in 17m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9bd337055ff4850a1369ba42594f2ee : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/586f305c55a642fea53a74a412bb7f6f : SUCCESS in 8m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4236a3b0a09a46b69b204e84483843b3 : SUCCESS in 11m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/52b3673493d2416c8e2086d5da1ccb51 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5c410a7adf724923b64939fc753560b3 : SUCCESS in 10m 02s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e1ac818174a3462d976f594a63f0405f : SUCCESS in 22m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/97ea727e84764c6191e7da40e83c16e2 : SUCCESS in 6m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d95de16d78e048cca40ef8ce2254fc1d : SUCCESS in 16m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fb5faee95c3646abbc7b812537de84f5 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08e2a113920345adb545390364de9164 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59004a24ca2c49f5b879a8e1dfbfc40c : SUCCESS in 27m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6c6c7e3f380548ebb8e2636dabaceac4 : SUCCESS in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c980f062c9234869b40de4f131142eb2 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32508fa0851641eba793a28e4f0af040 : SUCCESS in 6m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/216e816f882a4581b9f10cc72605ae92 : SUCCESS in 19m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d9727fbb72ac41538292ff986a97ba79 : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb594ca2750746149938b8244f54dc60 : SUCCESS in 59m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cfc496067a04ffea3914687deda651d : SUCCESS in 1h 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"15c1caea596a0346f97d3b189f7a008296a9ba1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:52:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfd12bdff1e2dbfd1148ed225d3c7d51ae61d23a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:53:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69e4548fb8bf46178f9a4b7bc6df5237\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a34554ceb794ce39de77781182222ad : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c59ea37b5f38ab20bb37f2675fb9bda85b4db806","revisions":{"c59ea37b5f38ab20bb37f2675fb9bda85b4db806":{"kind":"REWORK","_number":1,"created":"2026-02-19 09:55:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/977302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/977302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/977302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/977302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/977302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/977302/1"}}},"commit":{"parents":[{"commit":"b25e902b3f92b796e8c4527c288e6200f3a2a9ca","subject":"Merge \"update constraint for oslo.serialization to new release 5.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b25e902b3f92b796e8c4527c288e6200f3a2a9ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-19 09:45:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-19 09:45:36.000000000","tz":0},"subject":"update constraint for tooz to new release 8.1.0","message":"update constraint for tooz to new release 8.1.0\n\nmeta: version: 8.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I4f11ada8b0680d54f725c3398b345b910cebe431\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib55269dfb476eca64936bd7e4f60ac9d3e8b4c40\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c59ea37b5f38ab20bb37f2675fb9bda85b4db806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c59ea37b5f38ab20bb37f2675fb9bda85b4db806"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
