)]}'
{"id":"openstack%2Frequirements~977448","triplet_id":"openstack%2Frequirements~master~Ia040d10455b861f9806b35517bad0ac1ad3644c6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia040d10455b861f9806b35517bad0ac1ad3644c6","subject":"update constraint for oslo.concurrency to new release 7.4.0","status":"MERGED","created":"2026-02-20 15:02:43.000000000","updated":"2026-02-22 12:57:25.000000000","submitted":"2026-02-22 12:57:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977448-new-release","meta_rev_id":"3a4c1f8b673ead95bc0f272d5a0705c41eabddae","_number":977448,"virtual_id_number":977448,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-22 12:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-02-20 17:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-02-22 11:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-02-22 11:05:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-20 16:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-20 17:13:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-02-22 11:05:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"34e71159a9f4de4768ef3b93f96000f4753d4db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-20 15:02:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6006d148d9bda5116d5d58a77fdeabf8dfa8a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:52:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a9e7c160874128a839582f6ab82fbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4cc4b1c6466496eb6c228c8cebc028e : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6009a4bc38948d39e58ab7a2e5f268b : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/423dac12915c490d94d8b7e896e1e57b : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cb4f13b5fad48a28daca6c07fa4002f : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/951adb80855642169ba0110e5ad0f22f : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51d96834f1364c0396a3eb7d5a51f153 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea8c5056715f486dba7c0cc3e04a54a8 : SUCCESS in 1m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e41d7ba83bf04fb8ba044c7cf022a657 : SUCCESS in 4m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4126a9fc0b844bd485da084f8eeb66ef : SUCCESS in 4m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1372409bddf34a3990d637522eff8e71 : SUCCESS in 11m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/50538484ad0440edb312cfdc9a1965b7 : SUCCESS in 5m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a1208dbf11df4a2da3bedc591ae26e27 : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/056d6ef6d7944453a0df3d4eb70ca722 : SUCCESS in 5m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/99ec845966604fe29361e33cf7c1d2a1 : SUCCESS in 4m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ba5b8afea90848d2be86a114a8211522 : SUCCESS in 13m 14s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a2b8cf910f9d47fcbfa777f6a3a8f991 : SUCCESS in 3m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c5e946da2b8d44a5ab32c63f03660799 : SUCCESS in 3m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4efa964f03a34349b3985efe576e3a59 : SUCCESS in 3m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/36cfab0c1233424f90638b9efa851b34 : SUCCESS in 16m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e4fb7fd4b7c46a4b3431e02d9e3b92c : SUCCESS in 5m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0a087938379347118309b1e8cc5d0170 : SUCCESS in 6m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ceece75c1ac94906a408c76241bcc27c : SUCCESS in 13m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b6b4f0c3887e49a5aa15d240ebae13be : SUCCESS in 1m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/be5ec7f4edcf45219f61def13b33f2a1 : SUCCESS in 10m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2c4fe812d0d9439eaa0e36f69fb81b3a : SUCCESS in 3m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/481e09387f3b4de19781e323f02baf17 : SUCCESS in 2m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/125891457ea141afb87eb86e8ed8b069 : SUCCESS in 13m 27s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/312095b3139542c5a932e8a0b663ded0 : SUCCESS in 27m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/23f7f7ec84024d84bb8dddf94aee2de1 : SUCCESS in 7m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/51309d3fca364d478ff5d2b801cbbc73 : SUCCESS in 16m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c5811b186f3c43a5a887ad986957c652 : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69745d626e734a1f8faa14c6f8d37df5 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/decbdb4dceeb448fb84874de0f4bdcb8 : SUCCESS in 22m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b4e4e9878b114859998854824fcf006b : SUCCESS in 4m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cad49846469242a788dedaa334f66905 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94ae85efdb32495c825eb66c5ba832d4 : SUCCESS in 6m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/61797da8a0da4020ad8d6c52afbc8a74 : SUCCESS in 8m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7413240917414158ab9166fd67d16de2 : SUCCESS in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14193e30818741b48e4ab63eb4e09424 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eab922cb470a4afca4aee53aa61690d8 : SUCCESS in 1h 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3a36fe0d9dbc0cf09e54487e2f540c1ee3f9e8de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-20 17:13:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"642a974c97709fcd1420146d94a86122dfed305d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 11:05:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78de11483b594e3fe4224967598ee609c05abe03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 11:07:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23377937e87670beffdb47170db455dec5ed1e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:57:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405ed6d04683480e9b4e408a8fd23b16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fdc9b6068ae473c8955f57ec3629e0d : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce756958a8a448f4a43563c0f97db53f : SUCCESS in 1m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a42fcc4b066a48a0af4cc7c7eedd03b0 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ea8a90ec1a947518dce9200b6ce4e9f : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59ad3bacc48f46cc8e37e90513965d44 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67be84516fb7438c8e3e6c41ceddccf5 : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1ffe74636cb0462aaa150ad7fb975b3f : SUCCESS in 2m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8458a1f57c104ba983abccb9759a7e59 : SUCCESS in 5m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/909e1c921fbe43e0b14c501f4966b1b5 : SUCCESS in 5m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/afb0a5d8edb04a2393a04ab1d80edeca : SUCCESS in 12m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fb67056884ac4d508e16381078f0c2e1 : SUCCESS in 9m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f95cb3e8cd884450a63d4e13b4a7b347 : SUCCESS in 6m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7e0e256e77e4493099b48d0b5e1fc00c : SUCCESS in 5m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/32d825c7c90943f8b3e60f993882ff57 : SUCCESS in 7m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ad42837a1d1b491a800e8aade41283c9 : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e136985983c446a812f01e684413e4a : SUCCESS in 6m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/369986b12c2244e69b8c7622b71fd15d : SUCCESS in 8m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b99fe53173e640b8927531a6866bccf8 : SUCCESS in 8m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5da73d6055334b3ebfa100bf4a9f16b1 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/30fcd2cc7ecf40cebd11a8326e0a0a55 : SUCCESS in 8m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ea4c2926e8cd4b409998c9b9054b1190 : SUCCESS in 18m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e998122cac124b7f81e2859dc1b436db : SUCCESS in 10m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/57c682da21f4458e856b5f48b027df32 : SUCCESS in 15m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9fad5ae30b5f47b8a44706dcd24070ae : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0760a102a85843edbb827ec5be4c1044 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4308b041874a48c2b88cc98b4bc2c6d6 : SUCCESS in 26m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/01392c534c4b4c1fb111028ffaf1de8d : SUCCESS in 3m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4f16f7a684d340fcae1ea087267bb920 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbf09f89f33847808a38bbc6f33b6627 : SUCCESS in 7m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/888877aaa324473ebfe9c5cfa30f11a9 : SUCCESS in 13m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b3e0857ee5234c02ac4f3f7b8ff16cab : SUCCESS in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afd5105d6bd940eda9696b53d5cbcda6 : SUCCESS in 33m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/594bb9c7a2b94e8bb67d6bd2ac15e227 : SUCCESS in 1h 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3a4c1f8b673ead95bc0f272d5a0705c41eabddae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:57:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7a59ac2c56e17f4779ba632977bf723e7fe4f8a","revisions":{"e7a59ac2c56e17f4779ba632977bf723e7fe4f8a":{"kind":"REWORK","_number":1,"created":"2026-02-20 15:02:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/977448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/977448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/977448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/977448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/977448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/977448/1"}}},"commit":{"parents":[{"commit":"3ca33ef4f8971a017301909ca8ac0691f5782606","subject":"Merge \"update constraint for tooz to new release 8.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ca33ef4f8971a017301909ca8ac0691f5782606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-20 14:52:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-20 14:52:42.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.4.0","message":"update constraint for oslo.concurrency to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I4d7c44e00c54057bfeecf6141615566d6a7fcaa3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia040d10455b861f9806b35517bad0ac1ad3644c6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7a59ac2c56e17f4779ba632977bf723e7fe4f8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7a59ac2c56e17f4779ba632977bf723e7fe4f8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
