)]}'
{"id":"openstack%2Frequirements~977463","triplet_id":"openstack%2Frequirements~master~Ie366e8a81e6ac07aec9280082b107381304813a3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie366e8a81e6ac07aec9280082b107381304813a3","subject":"update constraint for ovsdbapp to new release 2.16.0","status":"MERGED","created":"2026-02-20 16:16:36.000000000","updated":"2026-02-22 12:56:31.000000000","submitted":"2026-02-22 12:55:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977463-new-release","meta_rev_id":"6a252cfecc73ed329829e0b70c4cf9aa924c8e51","_number":977463,"virtual_id_number":977463,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-22 12:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-02-20 18:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-02-22 11:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-02-22 11:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-20 17:53:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-20 18:46:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-02-22 11:04:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b66eba81d80a8bf708b8e12531a8eabd757ae86a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-20 16:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ebc32ea6eeb42509a39167f07c2de4e2512741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 17:53:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54212f2653b34c298dde9d2b19a52b94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bfaf65d1014370be883d12b716e459 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61ec7c52753412680a4289dcd49e17d : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e7863077bdb41f9bde65f0fb5a95a1e : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ba0eb87f0c240fa8dd2c856192ea926 : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2484991396ff4efe8182c1dcd78d2668 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b78d8c441e8496d899ecad518aa464f : SUCCESS in 2m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f92bd900f7a48d2a59b9dde37c68c2a : SUCCESS in 2m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2e461dacb8fd41b9bbe49f83fd5707ca : SUCCESS in 2m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/74481f795d6a4daf946c73a78d59b01c : SUCCESS in 7m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/91153656f8a14222a7fa1088373878aa : SUCCESS in 3m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/14ca566d0b2b406c9177c3ca916292d8 : SUCCESS in 3m 13s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c3fe04e27dac48fa9d774f7b4d5f51fb : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b78a6ab2d693462eb1ddc6b9a5ffa8ef : SUCCESS in 4m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dd46937915b340d6bf1467f66d24dd21 : SUCCESS in 5m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b117e63ffa58419cbded243f784e3603 : SUCCESS in 9m 43s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ff20fcc3c08043a5bd61dd8eb09638f0 : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aceb45f1f1414d5a8ea8606527f6f0bb : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e3459650bb664f07b8d0490a85de6fcd : SUCCESS in 5m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e58b155b920a4afc9e699868262a8cab : SUCCESS in 14m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/149842f1e7a14be9a0918ee7758b149c : SUCCESS in 5m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a67a648964fc407188d047a47820d493 : SUCCESS in 4m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/45f60db3e8154006905c2db5ace29ed4 : SUCCESS in 12m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/af2768cbd4d740b7936d4aa05e02b1f8 : SUCCESS in 2m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf8b21bedd64468a8fdebde5f199af56 : SUCCESS in 7m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/760a1f0805b64b2582d12e977a2f50a5 : SUCCESS in 3m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cca9375f55854b2594487f2320738bbd : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ce76153c97394f0594ac6bc59830647f : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cee2adb4ed3a41349f1a4c9df63a5d77 : SUCCESS in 20m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec08d94fd97e4c118148be6b399b07b3 : SUCCESS in 5m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9bcee5b5e2bc45c2960afd9f29914919 : SUCCESS in 9m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6316f9599f0d4594820ac09f663545f6 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/398cf67636434604bab69f4a21efc257 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a88f2a33b624292bcb5fdd5ff9140e3 : SUCCESS in 30m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9160b383e96c4743a61977f6486a7d9d : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/288b870a147d4c9c9720ba848205a7f2 : SUCCESS in 8m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30228490a01543eba24c78015a1892c4 : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/36dba4949dd9486ea0d343fd78eabc11 : SUCCESS in 17m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1cd3910d0ba44ae3b135e0c0288d83e7 : SUCCESS in 4m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa52bfa9159d4146984da7b156f9758d : SUCCESS in 50m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/036e205c548743eeac346e348e369fef : SUCCESS in 1h 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"237de79217e9d17f5a31f9774635a0f0b455b894","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-20 18:46:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa04ed38137881912afcce9330ffa252930283ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 11:04:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e86781bd19f1217d0ca04abc56c24323ebe54ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 11:05:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d2561ef7b0cb2873c8ba494b86864a524bf82d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:55:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e732f5fbf64e0cbd34960ee5a2b244\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5948d2b9be4c67bfc238f32492b675 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41672b0336394ce9b663561189d97605 : SUCCESS in 1m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e724b57f21834fa7a9a9832995a22821 : SUCCESS in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3defb548e3d344088e6c86e1e061dfe5 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8938a94f4abb45c698ec68e38568c808 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35d244b194f14787b681684513d6993e : SUCCESS in 3m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e47635ccd65b4b6d845b9df52752aa5b : SUCCESS in 6m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9bae8ba96cac4da5aa47dafdc2961210 : SUCCESS in 2m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8418ac9bb2c242db9d98fd7aafb48b34 : SUCCESS in 3m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5814201f7490437e95939c85c3fd1ad3 : SUCCESS in 5m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1902bac8b8c047dbbba7f6e1111202f2 : SUCCESS in 5m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aa4bdcc081e04ed59ee9d6a62e3585d6 : SUCCESS in 6m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/08eeae0eba5a40c7b44c145477146ed1 : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/44a15d905518444cb961d2001e2b7322 : SUCCESS in 14m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4643645e467941a08c95f43ec7b94303 : SUCCESS in 9m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6ce42efb943b4024bb9604bd0a86d35e : SUCCESS in 9m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/360d7ea065244513a828f61c295800f3 : SUCCESS in 14m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f69646919b5c43c89a264881febb8b10 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/202c8dbfd9d74febba3303c29eeaffa9 : SUCCESS in 10m 35s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/573f08e0a6f44823907651366148e728 : SUCCESS in 24m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2b35ff3740dc4146baad3b15acca1db5 : SUCCESS in 8m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5c737498c6a0450085ea20dbea76b8ed : SUCCESS in 16m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7435652e38aa43e7b0b6bce9443448e2 : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3eba595c8041404ca0ac96b00cf872d5 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d44805cf7a74c85be259a4bf5bbaff0 : SUCCESS in 23m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a35a3bda646f4264b293bfc9565dab77 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4f3d748d10bd4132b25335317e2e5bd1 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/918d99e366b344b3a7f352475ddd3584 : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/86e4cc7932064c5d80117dbeff622a58 : SUCCESS in 18m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2c56662042c84b3587ae78ed7e9c74cc : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baeaf9a1d4494bf0907b688d6214170a : SUCCESS in 1h 01m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2134b502b78f43c284ed62d1be8c13af : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"61440e7bfba2f777715c10f0928ad78202e313b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:55:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a252cfecc73ed329829e0b70c4cf9aa924c8e51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afbca6b18c44ba9a0b0835f20b979b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4524705b460c44ba8cdfd8d4861c12ee : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa42adb56b01879526879fde25843a86e5c4f103","revisions":{"fa42adb56b01879526879fde25843a86e5c4f103":{"kind":"REWORK","_number":1,"created":"2026-02-20 16:16:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/977463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/977463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/977463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/977463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/977463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/977463/1"}}},"commit":{"parents":[{"commit":"33bac846633a046569e3b01a70c8ce27ec2970b0","subject":"Merge \"update constraint for neutron-lib to new release 3.24.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33bac846633a046569e3b01a70c8ce27ec2970b0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-20 16:06:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-20 16:06:33.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.16.0","message":"update constraint for ovsdbapp to new release 2.16.0\n\nmeta: version: 2.16.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Change-Id: I28dc950efa8a4d1329ed2e6db0619157ddbefd1e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie366e8a81e6ac07aec9280082b107381304813a3\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa42adb56b01879526879fde25843a86e5c4f103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa42adb56b01879526879fde25843a86e5c4f103"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
