)]}'
{"id":"openstack%2Frequirements~977551","triplet_id":"openstack%2Frequirements~master~Icde1d76eda9c01cb52c1545c91976efd84b123e0","project":"openstack/requirements","branch":"master","attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-10-10 11:04:11.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icde1d76eda9c01cb52c1545c91976efd84b123e0","subject":"Fixes applied to automated upper-constraints.txt update:","status":"NEW","created":"2026-02-21 17:18:32.000000000","updated":"2026-10-10 11:04:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"8d527350f60d43774561c9a4df1947419ced554c","_number":977551,"virtual_id_number":977551,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-10 11:04:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2026-10-10 10:11:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-21 17:18:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2026-02-21 18:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 09:51:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-05-12 09:51:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-08-11 16:22:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"f4a8273a5bd8c38ef7b04c355fe5debe9b846821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-21 17:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db3cf068addeabe49eb386b894db190b3a55dea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-21 17:44:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efe0acfcd025b613b84359d9029c06d08c1fa7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 18:48:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4059f825094240c48250561e11b3dea7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65d239421594158a1f810ba19cbb665 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90568c096c7549feb598df00580d6525 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/516a37a85be744328347c66f2ab677b1 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68d9cf57f87d4687b89adfe02e3a1897 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e898cc30bfd47688734eab8ad500521 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90ccf65701c34222b75fb4b897bf6ca1 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c74c6db406014e64b85d8af2766dd694 : SUCCESS in 3m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f7b52a3079645f79046cf663c9fc120 : SUCCESS in 6m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b5107c4180694e94b30851949acceba7 : SUCCESS in 11m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9d4c12607e5141b6935953496296f76d : SUCCESS in 4m 49s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/739854eff9bb4641a105874b40224dad : SUCCESS in 3m 14s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3f2cbaacc95f4de4a16bae7d33b0ec08 : SUCCESS in 6m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d9da187ed9794a60b1516d65222b62fb : SUCCESS in 2m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9f4e23f9636c4fbba6829a0d0f162479 : SUCCESS in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e9e31cfece82453fb865e95c068f3d58 : SUCCESS in 11m 31s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f0110f61a28742b3890c25ff6ea2fcaa : SUCCESS in 3m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8f32058cdccd4671b02c843345b66f6d : SUCCESS in 6m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dd4c6c1af685476d9baf86322673e4d4 : SUCCESS in 8m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/82774a15f49b4ced8e7c961808dbc58d : SUCCESS in 16m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9630993c6fa54d51bf531e75927755f2 : SUCCESS in 3m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ccbde58a4815479d830cc6ead28c4fa7 : SUCCESS in 8m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/76f38fa536a54af7b61a79d10732213b : SUCCESS in 10m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/211241fae5ba401dbf7b98dc3286992d : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/777d76dac6634be6b89081b20db048f5 : SUCCESS in 8m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/76bfce3afd2a4bac8b5d8e13d922c608 : SUCCESS in 5m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/604e6df8e56746039d8c3cc5278d9f14 : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/12d3cb3e35194ae6a4c77a9bc03abc03 : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a2db3d296dc04ac7a238bdae22555278 : SUCCESS in 23m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79ccdb2da0cc4e3f9b45a1d9255b0831 : SUCCESS in 6m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/13f821d580984ad8afc4df1edf5809f1 : SUCCESS in 17m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/95b8e3f2a8ae4766921605214027c025 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d661cf46703041f1aacdb90e3888b0f8 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9948e0f2782e42deb508b7e0ff4423aa : SUCCESS in 27m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/136fb627762f48549dc92507154ce5f0 : SUCCESS in 6m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c830009a32754823a212c7ee9a0cce9f : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc48e8fb713b4973919327322cfd792f : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/205076d47af643f69900f0d11d7cd1a2 : SUCCESS in 13m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5393a67c98a3486abe7b06fb45ae59ef : SUCCESS in 6m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63d38e0694734344aba39592b23b461a : SUCCESS in 57m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c6a1d07c3134fcb84b462fd592bb77d : FAILURE in 58m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5fde5caec5b62c7b16cb87365b9b846fc7f1b7df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 07:50:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"633f219c00a6001f489e8aab01924437bc07190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 09:05:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5243bfa568a749f69c90752c1501680f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c5d856205f42fd9b0abc07b96b9b5f : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a429253f93444bb3946b1183d0c984 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5a4c23a122e4dd29c90724437842e06 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e917d7e0ceac4290bf0c30c9666286d6 : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98b431ef6b4344d4b3cdc5e965f8cf0d : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e6066088a834bf49d24ada23c25ba3c : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15f6782a02d34d7592dab7f16ad5dd2a : SUCCESS in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/494fb0430660423599d81dc1081a796e : SUCCESS in 4m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/eccadb460d5a41ef951a014018446fe3 : SUCCESS in 5m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8ab6b9f2ceae402da90c8378da390ee7 : SUCCESS in 8m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d2f76cfb80c14c3584773bcdb158bbe5 : SUCCESS in 7m 12s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f7a541baef0d4a5ca16947fbfbc1ed10 : SUCCESS in 4m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/39590aca4267460d89f3ff81fba3b823 : SUCCESS in 4m 41s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b46c7f8addf546e48de37e3b6162d8ad : SUCCESS in 4m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8128d2bcd74447bf9b6328e7503ce1ca : SUCCESS in 10m 01s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b19b409593e3412fa15e881eb831a2be : SUCCESS in 6m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/449db634cdaf4be78ac8dc3257f3ecdc : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7f1755718630436d85e52abd12577715 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5f998496362d4909bccf6340a285f65a : SUCCESS in 14m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea65dd69729243778893c898afed4625 : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/475a6db855ac448ea5b3b295d9ec43e3 : SUCCESS in 6m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7c7c256a6d9e43d9b5003e8c3923690c : SUCCESS in 7m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bcccf24aea1244bb89106de422b9b30e : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7d15aab5554d43f480efcc071f2153b1 : SUCCESS in 12m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a04ee23b5970476ab9e016a146e939ea : SUCCESS in 2m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/97a94b0f13cb45f389e7b04d6c08cefc : SUCCESS in 7m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/18b5bf32438a4e70903c16b98d6a97be : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7b4ff54f891f4080bc998a32e0bc4a64 : SUCCESS in 18m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3854c64850b64b5ebacfc3fb27eea999 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cae0fa0da40d4b22b1b82e086d92c4a6 : SUCCESS in 10m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2afa5be7ef104a71a2377fd0b4caeb5b : SUCCESS in 8m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85cf6ed288c54898b01265e801991e4f : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3514edf262e46bbaddd46e9b97102fe : SUCCESS in 25m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/15fc73f8e8544ecf83b16eabd731c39b : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9f2fc8a5cff644488c8ada0b46562146 : SUCCESS in 4m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcfb3806e8ee4c1fa1ea548ef70ee506 : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/638265da5c0e4a6f8bbe4189ed86999c : SUCCESS in 18m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fcf546fa66ba4254b71565f18ba6f57b : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21ee5ed54add47c3953552a5a5e37dc2 : SUCCESS in 1h 12m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e9b695785954ab897bd83f645611624 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":3},{"id":"3750efcd820f7cfcf3af490003a73dadf1ce81a2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 09:30:45.000000000","message":"Topic xy removed","accounts_in_message":[],"_revision_number":3},{"id":"21a4c2812f3f8afd2b291e86046807f7068d3b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-07 09:45:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"55a9c27d4c3d3ec561eb7b458950a1728f442bcd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-07 09:47:36.000000000","message":"Patch Set 4: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b98cb22ffa5bc15b750e999a2226c57abd194eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 11:11:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/224db9722fbf4c7c8645864302a565e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867506c12cc84a6faa3aa1f53f06638c : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f78487b7ae574ef1bed1ee5067c45c78 : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c794464ccb224524b053ef7dd3b96323 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d6c02eaf3c94219b29276dbd080d970 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7f6316d3b1e4dd1908038572cca63e1 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0cf44cb8baf41f39af29a95d6ffcd90 : SUCCESS in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a963ba4cbd1d415189f57ab34a779473 : SUCCESS in 1m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f3c9e180ba984f5997b49a81be91d668 : SUCCESS in 2m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e58caa9743e44606908a9899415de472 : SUCCESS in 5m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/26a09183200f490c9b071ad7f54b4672 : SUCCESS in 6m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2b1757bdceaa495bb39769fcd1f15fc2 : SUCCESS in 2m 59s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b437f143b852408bb68a494d8e626a31 : SUCCESS in 4m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b5d3820ec96b4f369447c3ef010e0998 : SUCCESS in 2m 55s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a3ca59c35b4b4d398ecb2797957dfa9d : SUCCESS in 2m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3f9571251b354c1086489c68705c7519 : SUCCESS in 12m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/16fb30407ba341249302c3bb1a05a4f3 : SUCCESS in 3m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b8fbc783d61f4a458284c7cd747f53bb : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/475c1c2286214191b289aae3dc605d0e : SUCCESS in 5m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/699f9fd23fbd4c42922faabe7aea65e0 : SUCCESS in 17m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/305b9a1c604248e086828ed07cb35e5e : SUCCESS in 4m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/419ff37cf4264786b1f927bba0788403 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2355044826424245a7103493ed20b1be : SUCCESS in 11m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ac4b589d3d4449eaadf813b29d47bd79 : SUCCESS in 1m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/671580ffda1a444a808b22ced8337cd8 : SUCCESS in 11m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/58cc608516f149ecb58c9d3deb1671ae : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dbe25f5f05914fbf95e1dfd48670dec3 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/172ee9b1424d4facbb67e57f194a3279 : SUCCESS in 9m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/aaf4a3e1d01041749ee28c1768872a45 : SUCCESS in 20m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9de4acfa3d664c5ba41b9814b69c63d2 : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7cc024627af840baa9f58b4331e4ae3d : SUCCESS in 10m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fad0500f66b3444b9649974c40b745b7 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be4a5f5cd1e444a99d76250c612475bb : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec1e08f2fd1f4d348afd2cc9556734b8 : SUCCESS in 28m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6adeeca569084244b2dff85c7a0a66af : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0aea8c8314cc4662a3adbfac389ada18 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa25fae8e89547bcac5b5c8249a6d25a : SUCCESS in 5m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/910c6ad9631c4097a9469c8e8c65a80f : SUCCESS in 18m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/914bd05b1b5c45b797756e9c2302f2f6 : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b381735de3fb404095e3076e9931f892 : SUCCESS in 1h 03m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39d75742c28342199db9e97fd67bc50a : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":4},{"id":"20741cb41f48c0484eff7c0ee65710bfcf29bb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-14 09:50:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f093a2a67aed88c7fdefe2561d3a6d4d86b605f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 11:49:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f746c030980440b087414e1634b7d56e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf48b0d4b0ac49109c6ee039138c480e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dad44f92eb4456985321967a5bf25c1 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5135ff25eca8400490da85e07e1b9716 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8992d7969e547ac8b77d295fd9aac16 : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0973dbbbe3ca4d1d8eb669be5f6b6853 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f67589bb59e4daaa1888aa3a314c9bd : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d097a9780f374b80a49b7e8e2fb534e2 : SUCCESS in 3m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e3b33c60cc514f899b2225d131a28c82 : SUCCESS in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1786dfc081824465a69bfa5164c6c195 : SUCCESS in 7m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/53a9d21708f64d3682ebce8cec05de1b : SUCCESS in 3m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d651ed8455404a72b876091172b3095f : SUCCESS in 7m 10s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ee56370c3f884e5fa516568ce105d5b8 : SUCCESS in 2m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8c5a2898a96643cda8f40e952d465876 : SUCCESS in 4m 36s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0cc880945a5b473b8735c68a589a853a : SUCCESS in 2m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/073d77bfd9e742cd9e0cf9ac7da6cd37 : SUCCESS in 7m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7bcde30e37d2420cb83308dd969b6004 : SUCCESS in 3m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2664e623af4a485791960731d93f7bf5 : SUCCESS in 5m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7f42ab92eb1b438289c44973c56127d0 : SUCCESS in 5m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ab82fea268c5417faa64ac7222e4bee4 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3101c13ae9ff42b287ac993132fbcdd7 : SUCCESS in 3m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4ebc55e72b724a1ba0d56d78e18a21df : SUCCESS in 9m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4fab36c2849f456e8ee006b0bef32c9a : SUCCESS in 9m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6c0129bb70f04bafa58c90f3b6b8ebd8 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/22f886f9a536496a88624ee703fa8fcd : SUCCESS in 9m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bd63ff9935da4cfd9ac55268c16a5904 : SUCCESS in 3m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4bede572aa7b4dca88a0cfdd3e5c334f : SUCCESS in 2m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7f4d7246efe948ba9c7adbe06ca13c0a : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ab91eef4c0244d929da4d4d456726923 : SUCCESS in 23m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ebf1a3928f6949b2a9749118ea1ee63f : SUCCESS in 6m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5c4ec2fb304e44a189a2caa9e6a7c323 : SUCCESS in 16m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6f4a5d8531e74fc8a3865a097d599b57 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f9d7696ccc843c28f88ddce6a13dfbd : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45160bceda1044b591566c2c6bf3cf85 : SUCCESS in 27m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/adb125e3e28e404e8a9e2040d5f6b80e : SUCCESS in 4m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f02ad300117641e3a074c48917ba389a : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74608c3d57664f4d81902dce46660d8a : SUCCESS in 7m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7840ec89ad8846338e46fdd0fa932c55 : SUCCESS in 14m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ef976a6aa0d449a2bb3a4ac8e951c807 : SUCCESS in 8m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/088cc52478fd4652bbda27481c61ace2 : SUCCESS in 1h 03m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3f31280133a42e8b53600c294ef9c82 : SUCCESS in 1h 52m 49s","accounts_in_message":[],"_revision_number":5},{"id":"31d1486ce5ff273165e2a284d14c4a4f8efeb325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-23 09:31:41.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"910333f226e22e03b58037bead835c00c26defce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 11:50:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d29a286136c48309dec5046966f947f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d98eba49ec451496df14a4e568e39d : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9803270a59489b92122b4ba2092cde : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66d980663fb64860bde1ec4a2fff7e91 : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c119fef542874319a70124ca8d443fd5 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/757e1e68b6e94e82a8d311e3ef7eaade : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/700b1884eca44d46a41c72c5d1209753 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/834469b5a66d4c12b24909b61a27c41f : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0ec33e24c9954fd587bad324892f0aba : SUCCESS in 4m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7ba75e828611478c9feade3b1d62cf8c : SUCCESS in 7m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0fcddd94db1f460cba0d75a9b4d193fd : SUCCESS in 6m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/82e79971fee646088d6e2e3bde73a8ed : SUCCESS in 3m 29s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/007fa97db4444f928170059065ea379b : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7d8fd008e42b4144925105d726bec692 : SUCCESS in 3m 12s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bd600c4ae957425e9a83e9f59189d17b : SUCCESS in 3m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c28de1b65a6f49e1a96e11d582e7834c : SUCCESS in 7m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/28aae2f084324a90b0049a9f3f6750c6 : SUCCESS in 6m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e6762c3c82994a958a35f48c0f367fe8 : SUCCESS in 4m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c7385e6e432f456db94b8cf6b22b17a5 : SUCCESS in 6m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4fc7974c8d5342eab802f0a1329a4fed : SUCCESS in 14m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c470555b1af42dbbf4fa5600cddb434 : SUCCESS in 8m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/60f77e356bb140d1bbdfd7bc899260cf : SUCCESS in 8m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4342105d60b8426daeb1bb3dc315e413 : SUCCESS in 13m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/783d749e7bc7435880e3f4d14aa0ff41 : SUCCESS in 3m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f8f783fea55c4432ab3abdd5661e46ee : SUCCESS in 10m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/41eba7dc729b40749f9a0d44341d6ba4 : SUCCESS in 2m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2d4966df09d045018bae5a58bd21a151 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/71777b78dc904d40b96a3e9f0f7835df : SUCCESS in 11m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3bedb2d113794af7b0baebe37ffd78c1 : SUCCESS in 28m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d054714efb7a4cf991773d65b982304b : SUCCESS in 4m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/458c2c49291745c29c7d9e32458bb57e : SUCCESS in 11m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/779b86d9fc2f43f9880c61f58af3f982 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e02529619e04d9ebaf268d778aab9e6 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49f334ccd1644f8ca0d5fb3ca91478d2 : SUCCESS in 25m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/46183dce791a4d8f96af4c36f65e37ca : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/025b6e2b59ce4126adf7c359cee59f0e : SUCCESS in 3m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a83254abb39942c5bdcd0de176b15bd3 : SUCCESS in 10m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b376bad025f1488086d6f1eb9f10fd8d : SUCCESS in 19m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c9ab42e59c6045f1a2d9a31be2a889cd : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0679b47ad1c64569a9998fdae6ee4336 : SUCCESS in 36m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/def97acd71434bcaaea2ff5c3b17b734 : SUCCESS in 2h 12m 39s","accounts_in_message":[],"_revision_number":6},{"id":"75cf1447b0f5c0d8c3ec5500ca254ace53d6a777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-28 18:21:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"48d18332d8399c9748cdd7cdcb2aa6cf8afa15bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 19:58:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6185a6935b476eada8d65375260c38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711ba3aa17414da58906730c652d0c6a : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4372ad4efe8425f9cb765d3bc31746d : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4386d75c8d3843908c83f983d79fc47c : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff63f4cbedc247f796bfdf7f31b44053 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c9ed4ed613d245e58f714d749801562d : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8735dd85e3bc4d34af6bcc2f6902635a : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a91941948c849578781db124eefa27f : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d818326caff4b7ea4a3272e98269231 : SUCCESS in 3m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff9ffdcbe2ec423c95f6011c3afc5b34 : SUCCESS in 7m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/90b84beea7e3446aa61029026878b6dc : SUCCESS in 6m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/705094510c6f4b42ae6daaaedc625233 : SUCCESS in 3m 46s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/850e3b34c4934274853251a6a6fb850a : SUCCESS in 4m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3b00315880044da78d5a2f3e5b92a370 : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28017df0b1fa4e73adbf58ec87cc6ff1 : SUCCESS in 2m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4015d23138664a4f97e7c1a25bf9f7e2 : SUCCESS in 11m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/95d73ce948f242089a19a4849463c19b : SUCCESS in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/35d850078fe547068b0fc017242ed7f1 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cdef663eed3c4efaa3fbae02b120c683 : SUCCESS in 6m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3c2571386294449f80d4ae05a24a06b5 : SUCCESS in 17m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b13677891c124b1c8d8d7fe1520d2743 : SUCCESS in 6m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/86331ae866a8499691da67a6cd36e460 : SUCCESS in 8m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/878bae786346469599e7e275fa6681d2 : SUCCESS in 9m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/274f4c2c969948488e16e96b8662f8b8 : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ff50812f4baa4e569dbdf1d8a97ecf42 : SUCCESS in 9m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/059b3b247a3c40b494b5cd2ec2688482 : SUCCESS in 2m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8edc8a44750d42c2a970145857357ce3 : SUCCESS in 2m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1af01efde3864802b53d9c47addbaa16 : SUCCESS in 10m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1540eeaa53fa4d849e7c30c78f7ecab2 : SUCCESS in 18m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/647cb6b5f82142c1b6fa4d5491ce9e48 : SUCCESS in 7m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3bc677255289429c9346e32387f3abb7 : SUCCESS in 17m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5b890eece0534ec693993e2aecc93a48 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/358e25c3d5df475980af01c9080ac052 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4dbae510c6346af913b10a7fcacad2d : SUCCESS in 29m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/70d041988e8c40b5946f54f7f202d06c : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b9aeec0ffcf146e3aa0c325332389faa : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9408ba20c72a469fad2d8fa97a0d5a65 : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f6a5f553bf284d44878a07328157f5c7 : SUCCESS in 21m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1b25e3c9c7744b0793663f45797ca786 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12eec192ebc04e20853cc395abbf75fa : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28635bd7f1664fd8b6dba4581d00bda9 : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":7},{"id":"7e9483ab14230afa0cad5954a01075fed8a93eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-05 09:48:32.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"69d7d12623d5c47d4e68dfc8c486fecc3a96bc01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 11:29:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024bd85554c7446593cb7b38d07ed205\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a9bf5653d64132b2dab67e5bb1cb08 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78e99d5aa69444e9ed02eaa9973e2ec : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee15b34948eb4a97a46fb93d0b7fcc63 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dc20530f40c4398ae74564bbdf3fe85 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/567afd9287f74be385500405989431fd : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/938a87a817514be48fa61c97511bd8c0 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f7c2900a8794416ac9f9a92c08c7911 : SUCCESS in 3m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6b19fe7fa4324097832107be33d7bf03 : SUCCESS in 4m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9ff6d1c9d8f04a39b816b8aaf01cc2f8 : SUCCESS in 5m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fa6cff6820e643d48b57f957b4daf788 : SUCCESS in 6m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b9a13c3e09a04f8ea5ef5dd7d37ffc71 : SUCCESS in 5m 38s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/40b3afb7ffe842c1bce3cd1f02b85098 : SUCCESS in 5m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cc5d2d483608402c93a0da245b03fbae : SUCCESS in 5m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3753f366e16a45d883bbe50585c698a9 : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/91c596fd943b4c51b7fbba6acc55abe4 : SUCCESS in 12m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/872c0d99604f44bebc69fb3dea38a729 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0ee311b1962d4143b36194b46276f2df : SUCCESS in 7m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3913c246fe3b463f8b3d72f25238b516 : SUCCESS in 6m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5db35c3dbaf2475f9ad75d0062374123 : SUCCESS in 7m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8215d12c675445fdafcf73caf1b5d9fe : SUCCESS in 4m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c0ba838815da4fc1a524bafb609af4d4 : SUCCESS in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/265a55ac65f741de87b57f40ea722209 : SUCCESS in 12m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/29f0d04c4e72405ab3c41fa9bc4a6261 : SUCCESS in 3m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/87e06204ef19401685cd60ff891791ef : SUCCESS in 11m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/093c0d8600f742fd99ba0974d54588bb : SUCCESS in 2m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fc6805fb37fe444f9efd40c8efa40014 : SUCCESS in 3m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bf595334eee1451484413151e1f430e4 : SUCCESS in 10m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/71c3d8ab40a7457f9a02c8f410b4f7de : SUCCESS in 23m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/55cbff9b926b4c8fb77d4d0eb8a62253 : SUCCESS in 3m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a21217a89a3d4f8ea44c951e89760856 : SUCCESS in 12m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2ca0b860bf284a2fb942d8b853852b31 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eac36c11bb1847c6a55c78cef952e404 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/028940a7540c4c9aa23a8f72905e8842 : SUCCESS in 25m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4330316cc96c4cd7afb795585f6a81eb : SUCCESS in 7m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/37403ce5c2e84e0db516f88d220a1ccb : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adbc4641374d4545aaef56c7409a6916 : SUCCESS in 6m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/89d23c617ff241498f15737f0a154c97 : SUCCESS in 22m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ad0b75a267e742ea86b4f3cf565819a8 : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fecf3531bc44cab89dd1cb261e903e7 : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2d36534b74546dbbf74f7e99301c420 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":8},{"id":"4cfab4ae548af454918040d6f70322f783df82f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-11 10:06:40.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"03ad9c208d5b12c9297b0c80571edf57a8cd0741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 11:38:16.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8c41b2cc1e94243b38d1b41a420e2c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3328f3c3e994b87893ad7315a0b7655 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793b9b7cb9884852aa793b27ec9f6dde : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3d93df7d058442d9d60bfadb8bbfd85 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84538426d4f141e5a3b8dccfe31850d3 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e64a16b3a5b640e4afe1f069ff6984ef : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab01927444b543988b4ae7dc527cda8c : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/599200b0509b4eddadba420c0ff26048 : SUCCESS in 4m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b3ad4259c8fb4b47ad7ea3b0212da3fa : FAILURE in 6m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c1b21075e1045ea81461f88d374c36c : FAILURE in 4m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/babb2e77464b40a5859a853f61f770e5 : FAILURE in 4m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/606bdf6a6dd54b61bc06228ac44212a9 : FAILURE in 5m 13s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/397affadf87047f8a6dcbccbf04f8c3d : SUCCESS in 2m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a203a022d0dd4b1ea452f0b142620a87 : SUCCESS in 4m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/acd1aebcf7944cff8acfd36b0e80b604 : SUCCESS in 2m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c5bc0f6a22ce4052bd8d94bcf9ab276d : SUCCESS in 11m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/52468834afeb49e0849a3c3b73c6b2a1 : SUCCESS in 4m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9b49dbd89cb948348cb19c0c34667900 : SUCCESS in 3m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d9d3639464a24ec499fb053bd87987cb : SUCCESS in 6m 27s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/4890e9dc1d7f4558843d184beb35ce28 : SUCCESS in 15m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c27a8708e53401492d5592266fc52a1 : SUCCESS in 5m 33s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/18d80962eee34e4db3cf24418b8a84e5 : SUCCESS in 7m 40s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a5312544cd464d9b9f04b00b3c75979b : SUCCESS in 8m 33s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e0881ff4b2454688a7e61045010d89e0 : SUCCESS in 2m 19s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/21b0a9e0c01a4ca5ba4a6e3ce6a413bc : SUCCESS in 8m 27s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4a7c53224ad74020aca5f84008e5dea4 : SUCCESS in 3m 10s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/cc40b1ed853a41a8a281e0f50494abff : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/54785f720dc94369a36a892e1d47b53a : SUCCESS in 8m 33s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/434bd40a5e2e4a699abf8d1cb7d44aef : SUCCESS in 24m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/979a50b078c84d01b7a8bfd95045eac3 : SUCCESS in 6m 21s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d2f6009118f14a868a9ebbc7a24cf126 : SUCCESS in 9m 52s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6e5217375eea40b792c09211c24e10eb : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edf282baaa99410a9e412e7eb16153e4 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b108887388cc4bbd9181ecf78651a7e6 : SUCCESS in 24m 27s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/282aec43a56c4fa2b3b581bd5b15a2fb : SUCCESS in 3m 37s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/732809b7e5c2410c8b190e655c55edfc : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15ecb71195234f8f86681dfc5afd8d95 : SUCCESS in 6m 55s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7b2467c3fc99490ca61b20274528a4cd : SUCCESS in 17m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5fdde16a6a1c4b3181ce8dea9040121a : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c88230c262ea4679871666a4dd67d9c1 : SUCCESS in 58m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26d7c845045a46f8804c444c1a1eba26 : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":9},{"id":"082b2b962c9eac5ad71bb09020e98b1d5685335b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-11 12:55:18.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a226c2a930d9c438a17b324c3665d60b44505328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 14:52:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1309b2d1ff743e0a4e3a4db01a1745f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3836c5bc72c4d99aef806c0fb4aa50e : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af2763e60f24d4ba99c43357176f780 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c41fa654d2b0426ab905497a07894821 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f167e08e7bc3454abc7b7424307613e3 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6bbc778da4b44c69837ed5f2c554298 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/895a45c411ae495c93ade863e9a8ec42 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b01e535c3f241dab7ee49a6853382cf : SUCCESS in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8ea91c11ae1f4f51815210228207c70f : SUCCESS in 3m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1733bc57f369447d8850bc4a475a0e42 : SUCCESS in 7m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/754ee123f82241d3915b041e62397cc2 : SUCCESS in 6m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67c97915c5ac46a88c2fc8088708ef78 : SUCCESS in 3m 26s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a26b05f76ccf48da89a536a63e59eade : SUCCESS in 4m 27s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/4429522ac43b4f61b969b43a228d2cd9 : SUCCESS in 4m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/a26cc3018ed8472888a3c9290574d1a1 : SUCCESS in 4m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3209a4cba49441b9bbb5151a9a3d5c48 : SUCCESS in 10m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a2b1e81895d842bca35aa64de306f90d : SUCCESS in 4m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/3cc571dc623641338d093dfa3e9791a2 : SUCCESS in 4m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/5e6d2113a4c24d169df54467052c2f15 : SUCCESS in 6m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/4a55dd40e5fc411fa1c69a323629fcec : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23880d8e726b4d33b67e47dfbb65fd6a : SUCCESS in 4m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2ac31294fd9a407dbc6323cb880877f8 : SUCCESS in 7m 10s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/93c949f711b9431782bf15ecaa7f1613 : SUCCESS in 11m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b8036be3826044829ce103e1442e08be : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c2b5c28d058640179e3682ff17a989c6 : SUCCESS in 5m 01s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8288fc80ab134d028f2f992e2fd467a3 : SUCCESS in 3m 49s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0221ebddc06d4b04a5e3ea1d0f816122 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/42cf8512d4404950babe796fd28f507a : SUCCESS in 7m 55s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fd76e8b7df8444f39d51eb5b980226f0 : SUCCESS in 16m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b5d2bc73dd94eed86b679f232afee7b : SUCCESS in 4m 50s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a0231b22b50148bbbf579159df29abab : SUCCESS in 12m 51s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/e898c6a0bb2a44928e02d88b177b2711 : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea9660fdc2704b699c9f959aea3431de : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b388106d7b6e4aa78f04c239c0382a98 : SUCCESS in 28m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/8a9d2d64091b4c0f9fad72c64ccf83c8 : SUCCESS in 6m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/670afa5c93664df89d0315a8095d972b : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12fe129ef69e4fee844b5337eb42906e : SUCCESS in 6m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f5c0533a7d824e7db1ce9160531deffc : SUCCESS in 11m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0bbef4d204b04a87be8af6fe23b9355a : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e81dec07081149e1b79ee4c6a36ff9f3 : SUCCESS in 1h 26m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c64dfd260d5c40ce951a9ce3ebe16ac9 : SUCCESS in 1h 51m 23s","accounts_in_message":[],"_revision_number":10},{"id":"5633d531a24af543d55534366ef226cda67c70cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-19 10:12:07.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"26f522c3857400805dc0cd8b011207ef2c84fddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-19 12:06:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80313aea964c416a9db95c4913cf7e22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc67753e4604e0da0c7f4573a04d626 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e275727de1448a864bdcf7bc6c42d2 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6b34f1a95d1432ab96b138e77f59efc : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59f43b8bdcb34999af08dfd6e6c6bd66 : SUCCESS in 8m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/784a48b8b4684698af407071828cbed3 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/566844fd7339480291ea031bc151d9fd : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b74b062219b1453899df01f25b9b7777 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/53914759ad104d8da9e533ffb8c6c48b : SUCCESS in 6m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b84da856b3f44dca8c41cd126a667fd : SUCCESS in 10m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6fd16825c13e4f4091e7dca1ded5cd19 : SUCCESS in 11m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/28f29af3716842699c1606ab19165b7e : SUCCESS in 9m 08s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/cc501f46a2384bbda603e7fc93ab09c9 : SUCCESS in 5m 12s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/1815dcfa5ac142078c5d44cdb350c62d : SUCCESS in 6m 43s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/94619fd29f2146b084b618456b0a7ec3 : SUCCESS in 6m 05s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/70b54acd41374e99996f3822d5caaf1c : SUCCESS in 11m 02s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/20915fce1a55460a9296ce0ebfa0b9e5 : SUCCESS in 5m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a0d5d521147a464d9193de4d75dc2d4f : SUCCESS in 6m 19s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b1a062ac592d435882681a7925a02688 : SUCCESS in 7m 00s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/833eca5f7ca046709c7acf7d73de04bc : SUCCESS in 14m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac734b6b071942afa2cf44ea7cb4ef65 : SUCCESS in 9m 24s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/f4dc803a3a6e4f6abd495458f509404a : SUCCESS in 7m 26s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3600d08a2d704bf1ae521e3bc5317983 : SUCCESS in 11m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ef1ae32cf35c442380c02ee78989e18d : SUCCESS in 2m 25s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/820e981ea3364f51875d3d37db380432 : SUCCESS in 8m 25s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0b30614a175c4f46b58a9e7c63648892 : SUCCESS in 3m 52s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/b94554d5f7da4ae7a08980bb7b28910f : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7b7ca8e5027446c5b91269547592f57a : SUCCESS in 9m 37s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/1179d85c27f34897acef2bdb7619b469 : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b2687c23714447389405df52bb9c47ae : SUCCESS in 6m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/37e44857e0d547d69f3e6625705dfb93 : SUCCESS in 14m 08s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/90274522cb194857a8e3ae7a4c40e52f : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c621a0e83c048649434e239998e7290 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/feae9da71d5b47be8dfa6b37f1272c25 : SUCCESS in 24m 28s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/14b201d642fc43758afb3ec145aac730 : SUCCESS in 5m 06s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/18341ac666ac46429d0b6032cf34431b : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b5160f2354e4adba452ac1dc6907941 : SUCCESS in 7m 43s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/fdb5645dcfed467f808d88bbea24e34f : SUCCESS in 17m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9ae2bd69e76346fe90e94b8859ecd379 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99fc4865525c4745bed11cb9bb709f58 : SUCCESS in 1h 04m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f44107567eeb41cdaa020622f970af86 : SUCCESS in 1h 48m 54s","accounts_in_message":[],"_revision_number":11},{"id":"5228fe093ab6034bc87af906fa642d2d99064c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-25 11:48:46.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"3840afcf762ba7cd418463912aa0882ea9f8a4c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 12:57:36.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bada4b6d3cd4f24bbd83fd93496f050\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee2fc3661e9465dbfe8b9e460db8426 : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdaf6dffb53e499dada252d837bd1f88 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9790e74588940eba6527dd3afb6cc63 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20001583553c4b4fa723bb8f909f6ead : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d72686559bf3477690a4ae6162e93c4c : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a80e942b80cb44e7adfca073dc20e144 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ba7809059d844e6a65065a233a1a3fa : SUCCESS in 3m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8387bbce32124a9488c17fce64bcac5a : SUCCESS in 5m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b4fc5396b1e94293aed74fadd6a0ea98 : SUCCESS in 5m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f4597407a98a4cbbb4a2f2aaa466e504 : SUCCESS in 3m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f0a73294b9b24859b3e4b4353954d737 : SUCCESS in 5m 29s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d7ec62c2b680466cbc653a079d863b70 : SUCCESS in 7m 14s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ca2de093db4a42f893e3c0849657ae86 : SUCCESS in 4m 02s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/073fac9b726e4972b68d1336c80cc5bf : FAILURE in 4m 41s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/1c2a44c726bd4970a4fccc9186b34f25 : SUCCESS in 7m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/fd0727056c314035992d0a4e609e8db8 : SUCCESS in 4m 20s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d8240d15ae1e429e8a8a9fc976c434c8 : SUCCESS in 4m 27s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b80d309dae5541e6a44c715563a0aca1 : SUCCESS in 5m 50s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/6bdfac7afc984bb0a75b4a445dd0ff10 : SUCCESS in 15m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/648b2e6a6df741098e7d7ebc1518a31e : SUCCESS in 6m 05s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6703021d512c4b159163191dce6d887e : SUCCESS in 5m 45s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8c38ada6701e4a3ea390deff5035fdf6 : SUCCESS in 6m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7c462aea70184641889dce48904d386f : SUCCESS in 1m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0eaad79682144dbda2879ad1eba29fb4 : SUCCESS in 7m 45s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/de17d65133664fb99d71b9414a9d1f4e : SUCCESS in 4m 21s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/63dc4c13eff744578ebb0da1d36d05fd : SUCCESS in 3m 27s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/81ec7bc47f6648e7b4c20c317de3b442 : SUCCESS in 9m 53s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4320628e835244f19222b39c1b3e49d4 : SUCCESS in 18m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9283cbed914644db8207bf4c78694ab3 : SUCCESS in 3m 22s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b30e1e265b1b4b5fb41092e0ae88bb9e : SUCCESS in 9m 24s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7112849ceae24c78ab3294084e037a91 : SUCCESS in 3m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5bbb04a7a81843f5a0fb908893514b78 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68859ddd673f48fdaef16adeeff0fe34 : SUCCESS in 26m 14s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/fee09ff5bed64b8aaca6b9296cb12cb6 : SUCCESS in 5m 43s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5e621fee2e5d41abb2f0484908b27556 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ed5c9e4c7d3475fb8b988a6108fd755 : SUCCESS in 5m 26s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/4be52978ae67493a87be1a901be02be5 : SUCCESS in 7m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7e46d73227884408aab62de3cf170544 : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1270707c1609414599fccc428ecce27e : SUCCESS in 1h 01m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0a6f7c4e85547ddb425f4bae36955d8 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":12},{"id":"0dd28ee02b513410e3f58de39179c1f96e68d59e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-04 08:17:34.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"d0a0f1b6c69b0d6974f0b0452c0f45dcf844c330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 10:03:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15c96f65ccc441b5a3262f594584a470\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9868d73030494c3cbbfe8b629185c1c7 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e6866b0ee8478d894d3c7bb89f08e1 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bd3acb9761b49418e6b57b265817542 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61785fca25894e71bb1800934251622a : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e96802c222b4af7a008efd430402c06 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39636499cdf44c228d92738bb4185109 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f68d62025934309a528105085a64292 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/94bf7144474a4d7488b4985af53091e2 : SUCCESS in 4m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3d078e458284425a420328528f58307 : SUCCESS in 5m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/34787f8bbf10457383625493a5f4e309 : SUCCESS in 5m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a2322716386d4e53bd29a7652469e255 : SUCCESS in 6m 13s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c69873f926a84119b9b72d92dd74f7dc : SUCCESS in 6m 16s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/e310ccc5415148ac84762162cb548c82 : SUCCESS in 4m 53s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/37582dc436ae46a48825b5d137d5b1ab : FAILURE in 4m 32s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/2f867a2a93af4d0fb780d019aaf78fee : SUCCESS in 11m 41s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a6e75adde4f247b485b5b4008ea04242 : SUCCESS in 6m 23s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/cb01b537b3de457aa94f9bc1d1d75042 : SUCCESS in 4m 51s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8ad14007022d4216a8f7c5416544b3a2 : SUCCESS in 3m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2ca14bdb5eec47efaebdc7bd2277ee14 : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a82100916094e02a550effa1ee1a9cd : SUCCESS in 7m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/d965f6edde994050b05cb5202570e26a : SUCCESS in 8m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/03ebf6cd125d4ee1a52f867c7a7acc4b : SUCCESS in 10m 46s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/ea09750784ca4f62a3e3cd055c0aa954 : SUCCESS in 4m 42s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9f9d43a851404adf917c651e774eefec : SUCCESS in 7m 58s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/33ab4b46a89b405b8c3e1f9b1b739e42 : SUCCESS in 4m 38s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/169fa75216c3491f80f74df923042b0f : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/c3e4757e9f9f493eb8f5a107df576158 : SUCCESS in 10m 26s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f9dcabffbb8047ec862150fb7812c2dc : SUCCESS in 17m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d46d17486423420981c6c96092a12ee2 : SUCCESS in 6m 51s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/49dfee551faa4d3bb8e3a6f6d8f59687 : SUCCESS in 13m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/130846aa4ca644c3b6d8a2d1ee36590f : SUCCESS in 8m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c96aa18fcf0b4963be894a4e1c20b228 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3afa027c16894d1fa5e6bf840edde489 : SUCCESS in 23m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5d288d8dd39a43b1a56fa8296ffdd69f : SUCCESS in 4m 50s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/2a5e8e1342614b57aa9e9dcc3be77e29 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d7325f4df5142fdbfe25b4cd8f766f0 : SUCCESS in 6m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/96e1a23e51194a18a74555efad003698 : SUCCESS in 17m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e6ecd4405c0143c89b5e88262065c2ae : SUCCESS in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31c077ed51c24416a0b689f24d17a417 : SUCCESS in 1h 16m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/998c71df026c45bcab69d75972932a8c : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":13},{"id":"1511962336cd8ede0abf85b077ca5ad07de3b132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-05 07:28:57.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"0fe46a439788b04315a9c49bf028b25a51aaa7e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 08:52:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d9305dfa334403989c19113c8aeb34f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bee63c545b34f229366321e8135cf0d : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb8efa569214ba281abc95286ce3f93 : SUCCESS in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ce05fc1aa8d4fbb9f22d09e8009f7e9 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/307bca3b2b04485d9bbd9ed236efd1a5 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13da120dd41140738962faeaae8fe7e7 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3508193f767c424da44a00f78f8af081 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97d644869d5642ed962af1d09f6b5333 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d810beeafb22497ea45b4b1e66ba372e : SUCCESS in 6m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f184bdac71d4ed9ad3e503271451092 : SUCCESS in 6m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f141f86a1f9647149dc5c6a3bfe08e85 : SUCCESS in 5m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/027eb0266d85472b9c5b48104e0b0415 : SUCCESS in 3m 44s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6f3397ef2d0b4ee3ab2dbd30f9ebc2cd : SUCCESS in 4m 10s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8dc8093243c44bf6a0bd088ea964dc04 : SUCCESS in 4m 39s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/fe2a3d59214745c89f237816b9f33b72 : SUCCESS in 5m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/298bb28de99a4964a33d2d948a2d4ed5 : SUCCESS in 10m 18s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7f2aac1bdd5b4f139487ae772ebc5e07 : SUCCESS in 5m 14s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/44871f5d9eeb4b84b19a6160ee0d0d37 : SUCCESS in 4m 36s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f0174ae772534c71ae8d911325425d28 : SUCCESS in 7m 11s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ef69362ce489478a98f45c5c65e20526 : SUCCESS in 6m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5edf8c35412a483cb9cd03d876bc3a29 : SUCCESS in 5m 47s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/da739bbaaf734dfab818dd5a943c353b : SUCCESS in 7m 24s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/114eb134ccdd4bbfa769e16dc0af82f6 : SUCCESS in 10m 35s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fd62d65004914c5e9f1dc37e8ce64c39 : SUCCESS in 2m 29s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/bad766e482644fd5a1b232a1eede12f0 : SUCCESS in 9m 26s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9467184252d147d59246172bb0fae302 : SUCCESS in 4m 01s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0862a03caf0049cdb37314d052595f7f : SUCCESS in 3m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7ade6c7b171a48f9bc6e2c63a44d3e63 : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/09493246e83548d99705edb1a3cf2ffd : SUCCESS in 21m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac231358f00145e9ad3b33ed00eddc4e : SUCCESS in 7m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cb465ca12d784f56947060d37477c1ac : SUCCESS in 13m 39s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a46cba0f73cc47319069ba15d984b127 : SUCCESS in 4m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/850cc9601b1b4a479c367cd3e566c302 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8988665da564640aa6ee28c0df243bb : SUCCESS in 20m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/a3f9a1931ff94003ad8cb6d48ddebcd1 : SUCCESS in 6m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dbecd97cfc2a4b3192c3e24dac9e56e5 : SUCCESS in 3m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/623677ccdbb34fdd8c377a96e980935b : SUCCESS in 6m 10s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0ac61cb940204096b5f92e818c8bfde6 : SUCCESS in 16m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1b8374fedb7142ffb8e981fe474b3c77 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c7f7b8789f5407f92de97fed51e0cd7 : SUCCESS in 1h 02m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4119a1c7178c46edae8f4c69fa34c628 : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":14},{"id":"22435ddee150add80d77b54d3e74c2256db76957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-10 08:04:28.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"d4792290ed16f1e381d772b0c9f0f0bcd2b24d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 09:59:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc93b4dc854498c85390ca5f6e21e84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019e008caf2f4beeaf6a3a778ce3b5c5 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a605a93a41fb4c30b8c7f12db2798c4f : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95cfde2d2e6c41f7a5645aa36080cd42 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0b2727fc1b64e0b883215faef916488 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c69959b8edd4cf38d5f443a9f4b8b0c : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d45fdf1c6182482d841fce52671334d9 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/120e87af1f9146cfae0d9252639d8ea0 : SUCCESS in 2m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4d688212ab5746a186214d80d78dac12 : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/575dc07f02dc46668efd300627733075 : SUCCESS in 5m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c9c0d4958bdf4ec49e85925391d7c10f : SUCCESS in 8m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67b544aeb07c49e6bf55fbbe1e016301 : SUCCESS in 3m 56s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b67be81cb32d468a92613ccd5edef8dd : SUCCESS in 4m 34s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f04200abb35e4610a4f588a89e778d53 : SUCCESS in 3m 56s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/afe3244ef0cf40b2bd3559fbdf6bf995 : SUCCESS in 3m 12s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/02ab857d667d4f4aa201c7bb5d235733 : SUCCESS in 7m 34s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/37cea68d59bf4bf7b5855f7808a6ebf2 : SUCCESS in 4m 24s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/61f482129dbb45d9bfe49f081d3da7e8 : SUCCESS in 3m 51s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/b5d01f902b224ff885f9efc84c6e7712 : SUCCESS in 3m 41s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/80ac1ee2d0ea4245bc6043227a872b87 : SUCCESS in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1ca9778fd19406e87e05d1f0f9f04f5 : SUCCESS in 4m 34s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/cdb3c9ca04c14f7ca24fdec0239b82d3 : SUCCESS in 6m 40s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9c8780e9948148939378728bd8dba458 : SUCCESS in 7m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/043ea6ac08b2474f8ad7edeb063cb78b : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/059975e9603543cfa45c84abf0a90b18 : SUCCESS in 8m 09s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/1981c06917f84899a84478c691035724 : SUCCESS in 5m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/66054b6001c541a5acf90757bdf2935f : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7cf3976ff6304255b787870e2b75c35a : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f4b2b2b5be7645f39c65f4c6381de4f2 : SUCCESS in 17m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4dfa26e999e445ea9125d6af0a9a6123 : SUCCESS in 5m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/c48bf594301642b8af4cc38f2d9a7da2 : SUCCESS in 12m 53s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d0a724fae5ee4c1fbd0a39bfba6b1b9b : SUCCESS in 3m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9095fd96ee474b11a61150d8b5c9afac : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b39e3698ba24963b92145b522b12eb1 : SUCCESS in 23m 42s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/22b0889b801a490cac93cc837b90f758 : SUCCESS in 5m 26s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4383f13d33734ad49cc2dc1830d53b43 : SUCCESS in 3m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2f736645cc644aa8a02dd07581765bd : SUCCESS in 5m 54s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6bd1dc23bf6641f6b119ad552a63c6cb : SUCCESS in 15m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c0721aa0269f4440a401400df1a7c08a : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41674fd919be4c6f968af2062f8eb9df : SUCCESS in 45m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcd633e336354541b5f45cd68bc6a3f9 : SUCCESS in 1h 48m 44s","accounts_in_message":[],"_revision_number":15},{"id":"b7a25195d7963003b634e6ec1f08076fa0a2785e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-11 16:02:43.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"62bc5790cf7d90124879ababfabcb6b537ca25d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:49:41.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ff4aeda74da4d699bf0f87db0d84f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96414548cb6b48f0b4019e7535104d3c : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/362c26879d5e4a0daec065fcf605924c : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c8e34247cc44c3c98b65e531e870398 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3431e2c80bbb49b6b6be5733cee130c7 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61f7545ed8094d5d8bb93dd479dab751 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83380fdcc0c94bd4b1131106470fdf82 : SUCCESS in 2m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c92411f6c29f4dfbbcec42370ea0f037 : SUCCESS in 4m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/52745365efc042f5a66f8d89f2272dda : FAILURE in 3m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8ffc051388a84508a101af9899aee8b1 : FAILURE in 2m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b3acb61a141e4a9aa8b7b9dc24425cf4 : FAILURE in 2m 37s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7a962db8a7164380a3e2e0ee0d24f4db : SUCCESS in 5m 11s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a28634f52eca40d49be131fefc7ded0b : SUCCESS in 4m 39s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/0f6081fdba3e486d9ebcc018aaccc3a4 : SUCCESS in 4m 45s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ee6908765578483284457a57caf26367 : SUCCESS in 8m 19s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/81cac9028062441faf2468a6eab17126 : SUCCESS in 3m 31s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8ed528a252654efd96556fc10bfa4b60 : SUCCESS in 4m 32s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/138e348ed4644bc2815e3946c4f99007 : SUCCESS in 4m 19s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/de32447daeb44d99b38be2c60bb62dce : SUCCESS in 15m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6e39a641df440ddb3c9a601f365e176 : SUCCESS in 6m 45s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2c8a8fd19e864adb8097ef2a1cfc7c93 : SUCCESS in 7m 35s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/7c0b50bdc96446cd9907f010f46bc403 : SUCCESS in 8m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c5c6f527f3bc4d80bf0fd575ac9a65d0 : SUCCESS in 2m 59s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/783b522395dc4e7ba3eb26aab70eed55 : SUCCESS in 8m 32s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/edde273d18964df882902a9a11f70246 : SUCCESS in 4m 17s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c087cd2fb63c485999c7f07de1d215a7 : SUCCESS in 3m 01s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7c47960d46394f47bf3a42593ec8149d : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/19ddcd5ea4ea482689cd921ec8d56a32 : SUCCESS in 16m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1540ced7644741b0b25c0af65dec47d6 : SUCCESS in 6m 32s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/a674e9a327c740d4983e7f4e3c93ef19 : SUCCESS in 13m 23s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/f5ae52df23a344eb8674b578b869ad7c : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22ef25075c60456eb659ae53be842f2d : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2d86d67b97640f295a67048b470b3b5 : SUCCESS in 25m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/79018ede7d284c44b32bd80fbdf6da0f : SUCCESS in 5m 59s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/d6362b482b644c4683c1932670643a81 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8bba2218ffc4a609c3f4992fe269a31 : SUCCESS in 8m 18s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/96145f7423344be0ba18517c96a75b20 : SUCCESS in 18m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/76ea8d1da72f45d0b75bd13a60918578 : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea41ad3c2db144d198133cbff13aca46 : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85bd3f39416149f7b7ac0ee5d49fe1ec : SUCCESS in 1h 40m 41s","accounts_in_message":[],"_revision_number":16},{"id":"809162f522e1f89d077e8ca66a182f72e8266de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 07:22:15.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"4e6672e6d2c482f392c63def853ff81aae8d6519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:02:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6fd8416d0643fd9c89f2e58efaf2ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b62a59306c43a0ad9c46aef54408fb : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14634e283e6d404fb9e9016c17508238 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14a453e536a3463699e6416679899d6e : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d76244b6014642fab7e5f3ba16900b62 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/68ebff1c767140368fdc1783fb69bebb : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3479191f61d246159b889165e11a3181 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a0c2c9a09ee44c69df8bd8f1ad6b9ce : SUCCESS in 1m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4f2b3f9414474bbcb9b20d426f1748fc : SUCCESS in 6m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7837bf11ab744a3a9cfbbd20237d58d6 : SUCCESS in 5m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e682f18c9ae343688cc4387001dade22 : SUCCESS in 7m 50s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/01a8a0d5b5834f5fa8d1defc990f6bf1 : SUCCESS in 5m 58s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/f8a397fb701f415985fc0d993dda29e5 : SUCCESS in 6m 23s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8f3fe1dfb08440eda62ab9fe3a09a3c3 : SUCCESS in 3m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/9b22027d36fb4e54ab95b46652874879 : SUCCESS in 10m 12s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1bd94964b9584626b3b6d27261db1339 : SUCCESS in 4m 32s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/42edbe94ba6648e0b37a5b571b8b51ea : SUCCESS in 4m 49s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/de204ea1ad6f4c899dc4f97636859b21 : SUCCESS in 4m 42s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/31593caa2c6d4a4597bdcb1af0c1b4dc : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ee68d83484c49538937486738057eb9 : SUCCESS in 3m 53s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b54b3caa2c2d46ef84d5f83ab9481141 : SUCCESS in 8m 16s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/15bb67aa86df411db324d87206d9c471 : SUCCESS in 9m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0d2aaa4d071440058cf07d99ca9ef1c9 : SUCCESS in 2m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/30010fabdce644df9d99c99ddfcbdc37 : SUCCESS in 6m 50s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/fe0d4bacbd384110adf683c12c00b7e6 : SUCCESS in 4m 34s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/3c3180a3d59443908972b19642c2dccc : SUCCESS in 6m 05s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a015c73617af455794e40a10ddb5bebb : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/906055733d414adaa3ddb388231b7b37 : SUCCESS in 18m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01f899f8f01c4ca698c4a96f65081475 : SUCCESS in 6m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/26d989a4e841424fb851ca8a72ca8bb5 : SUCCESS in 13m 39s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1ee6f145541e4556ae83096461759a18 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ac5d77b30d44077917f7407a17de4c9 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eadea402c71645c7bd4d2f3ae4dfb628 : SUCCESS in 24m 06s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/40e4a3c719464d9db09c6cabced86d10 : SUCCESS in 5m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9a8b1acd9b50415f869942c08b301778 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30aab328dff241aab32f868c07c777bb : SUCCESS in 6m 21s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c562cfb999b44c63bd611edf7917a7af : SUCCESS in 13m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/51ec00f861264a83a253cfa92f7df445 : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70b2e7edc64749bfa2c0e24eca5e5f9f : SUCCESS in 1h 07m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/903cd2dd93da42fda52faf4e79c57551 : SUCCESS in 1h 33m 22s","accounts_in_message":[],"_revision_number":17},{"id":"e79b73f4d02b6bccb4269308f93952c2491a23cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-16 11:13:01.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"dca3768c9cb7ace6112d12a96351a640ea8411fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 13:03:44.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d740b9db96a4684bb0e192396aa2f6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b46fbea696e447c81a8cc16394e4c24 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b897f6953b4d14a4c2127f0bc7ad6a : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0105b1e63cf45f192b996b9f2e53bea : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc7dec88b98c4afb9a79d56b3b85dcf4 : SUCCESS in 2m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0508d06f4e6d4439b30b2cc0b3c6487a : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/241faad5cebb44ea82743b9b2facb8c8 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c179454429ef43e39bcd4da205d0d304 : SUCCESS in 4m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ee105f91f251482b8eb5b869ca9affaa : SUCCESS in 5m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1973e67c31934a9abe6744ff9e2b4380 : SUCCESS in 5m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6a5b9ecbf22a4993ab1af16829d8f304 : SUCCESS in 3m 48s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/18b98cc689944866af04bbbc6c4a7948 : SUCCESS in 7m 23s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/a1417642649e4b81b25f5bda4e585372 : SUCCESS in 6m 07s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b48cb35f5daf476dad2e50e117898d1a : SUCCESS in 5m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/82ba806e31444a80adcee3325b733bf1 : SUCCESS in 10m 50s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6964b932eec349358f947bf4db676475 : SUCCESS in 4m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e81e05e3e9f84650a49cad1d5ef86c9e : SUCCESS in 3m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9c9c814f9a1142798532032897d8d020 : SUCCESS in 6m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c5f0f4b2709948abafbfff89ecdbdefb : SUCCESS in 7m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75a731e073da4cbca20f1af1258a3da9 : SUCCESS in 6m 56s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4b1597d71714437c965b5ae13171c7b9 : SUCCESS in 6m 43s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/60c576d3ff114ddf85288678bfaa9fa3 : SUCCESS in 10m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c84969bc9b96436e83cad24c6023cb7e : SUCCESS in 2m 36s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4d87f956390d4ae6ab1ceec883e936fa : SUCCESS in 9m 23s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/08893fb07f6f488c953ce14c2799884c : SUCCESS in 3m 37s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/9aa65e54aa264ba3b505df60a89df012 : SUCCESS in 5m 51s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/ea370c0a24164fe894e18a824bf0878a : SUCCESS in 9m 28s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c4e0e9246cec40b99600698969554e4a : SUCCESS in 18m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3a33fb78c39040d9b0544a04cdfe8d09 : SUCCESS in 5m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/593d0e0318b24ac389c74533923116d5 : SUCCESS in 13m 33s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2ecc0c6950634102b96db381813eb771 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f1b97db8c064c11a60be265099ffc61 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4292b672cbe54d48b252b63d6a3c8bde : SUCCESS in 24m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/55ffadf0a3ca4ab1933c5f1260d7905d : SUCCESS in 4m 32s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/8c521c7c217c4acc83a0bf7816b8109d : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73bc964b75e24e268493bf80aef92aca : SUCCESS in 6m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/26b9b181a6984c6aa60f72f0bec63f4d : SUCCESS in 20m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/db4076da6f6b467699b80f903e7a4074 : FAILURE in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aebdd775906e47e08c6ef3e3171fb711 : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94bc249aa36b4c95910147e1fdc72d5e : SUCCESS in 1h 44m 32s","accounts_in_message":[],"_revision_number":18},{"id":"d957590f03bb9050c002af4c4748eb15b1ff5854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-18 17:17:33.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"2ab865a6d9ae2ac681c8720af438ef200f82117b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 20:24:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54823c5f2c5a4e83ba3ce64c99c6e7be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418b0b45d9fd4395a5bcd8d0caadad9b : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3440cb4cfa6444ff8b9c988aa04c82f3 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9947c8ca08a24727b8abdaa2ff3edbb4 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/508e59673c244e0997d620e5109a8339 : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/08bdaf1d3beb4268b9af4cb7bd3149bf : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af21b27c35424aebbeea82423f764dfe : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0591ea2aba6456f981fe0fa8678966e : SUCCESS in 2m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1737bff2350146ee9343bfa662e2e4e0 : SUCCESS in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/72e7eac7e0004b6f883a9b9461cbc5ba : SUCCESS in 6m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3fb4432f22674f0d9c3a4db4f7601d23 : SUCCESS in 5m 41s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/07355d97541a4482a2e8fc48ef593ac2 : SUCCESS in 4m 47s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c7a54a6f06d14ce7ace1517ae80e115a : SUCCESS in 4m 38s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/fcf524ee4da249e29bc2442d94548ae8 : SUCCESS in 5m 21s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/38a2080352a14698ae80ca77f6796828 : SUCCESS in 7m 44s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d15d71313e6a465db0edbb4f80cf1843 : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/738a627a97574b7596279a5b2c9e7c6a : SUCCESS in 5m 28s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/eea8e5e5d381495881d7a1cc0dbf5ff2 : SUCCESS in 6m 08s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cf94699dede14b58b4b031b8ece3d565 : SUCCESS in 7m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7498d0439d234942b5fdc4b0761a73be : SUCCESS in 4m 23s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/2ab673bc56504596bfc02651067414e1 : SUCCESS in 8m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4175961ac2d04315aff48a70f8a9ec7c : SUCCESS in 10m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0b84f64b3a524355bc740b59a8dbb7fb : SUCCESS in 2m 39s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3fc36b1c244845568cd4638a7b323f12 : SUCCESS in 9m 14s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8e77e1d71cce422b9e21a43734934840 : SUCCESS in 4m 26s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/346196ee40c94f618a9464113a37d174 : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7c57b551029c4ce59fb677d6228f5a95 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/cabce836e07f4fdeb2346152a74c34ca : SUCCESS in 17m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/137649ba5ada4762992918ea92f15606 : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1a621d619dbf4a17b908306894fadcf8 : SUCCESS in 9m 02s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/34a409d8e3984b3ca0991089d2849177 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de129d4d50df4198a15ec5fa44d32bbc : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/920c59fb76a44030aa5cb27ad66296a2 : SUCCESS in 24m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/cc059bf73a214fb9a8f791cb5b7b7d8a : SUCCESS in 5m 50s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/045b68df7ccc47468ff45bb6ebc68d6c : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c63f654ecfc40379e897a87f541558f : SUCCESS in 6m 54s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c4827a7754e8423c846781658f5146ae : SUCCESS in 18m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fd8b75d269c74a0b84e3b2b7f7f9226c : SUCCESS in 7m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b89ecab42e0740d9826e2ba504de7959 : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c01eea1d77b24c2480a8ab17fe66993c : SUCCESS in 2h 16m 21s","accounts_in_message":[],"_revision_number":19},{"id":"e93f123cba83668c1218d3ef3cfd2f034020ce9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 09:49:37.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"83cf83add874b0cb68450dff83f4df9db7f7e8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 11:34:25.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dde946dd41c3440bbe598ea5c787dfd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b973b2e5cfa4478c96beb513c3125388 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9088a6ed614f7ea33d32bb5ca9c341 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbe6c7567f774a03b796ebda2ad3e91f : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f20c54091432491baa54d36f0f4ceacc : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/68bc9596a023480c8f17a490e7f87033 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0003b657a5574f81804bebf70308b3fc : SUCCESS in 2m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e47433b2816433f964109c3d9fcb1b9 : SUCCESS in 2m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a10d448dd33241439ca0d7ce86855bae : FAILURE in 2m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/eb894d9896e24a87afede82dc3e7749e : FAILURE in 2m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/584dc0b65a664560abea9d1befb1db99 : FAILURE in 3m 37s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/1ae79646013b4f4689061ca09a105f97 : SUCCESS in 6m 09s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/2bc27a9cc4be4186bb3f60178088b829 : SUCCESS in 2m 42s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/9a4eedb3520349648bdcc53719ce576b : SUCCESS in 4m 42s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b8fa2163773746949ee9be6427738495 : SUCCESS in 6m 22s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/41efa1195eb94782abe51b7540173bb7 : SUCCESS in 5m 10s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/7edd7faaa06a4ae0ba17a2147d9d2051 : SUCCESS in 5m 28s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e658ff60da6740f2b69f689a421e9efc : SUCCESS in 3m 58s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/90dc7d4f61de47f7b1cfa65185b9edea : SUCCESS in 16m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c492a8769cba46fbba1dec6117630252 : SUCCESS in 4m 49s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/20ba88ad13df4e78bed3223469bdeca1 : SUCCESS in 6m 55s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1017099a99fd4521b9014f5bdc496fdf : FAILURE in 3m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5b629f408b5b45858c5f284e2dcfe7c5 : SUCCESS in 2m 19s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f18228f0ad4d4937b9b8095113660a0e : SUCCESS in 7m 44s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c959e397c9b0415b9833fef38977327c : SUCCESS in 5m 42s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/702c4d3074814190bb447ffcef9d56c2 : SUCCESS in 6m 25s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a07b4cce6e8e4b75bc582ed19285db1f : SUCCESS in 10m 13s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f5c816e72eb145ae8d0197bdd273c75c : SUCCESS in 25m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d771cf65bb114c8c9f33b3fe0abafe42 : SUCCESS in 6m 19s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1bc3dff7dee64b00ba19350a64a4fcbc : SUCCESS in 13m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/00284cb262e04f0781854db79fd4b8b1 : SUCCESS in 3m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93877a7136fe491d89682f558838cb6b : SUCCESS in 2m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7eeb41c42b1b4ec8ac0e687f9bad993d : SUCCESS in 23m 06s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0916cfeb4f874ea5822dd83732c23faa : SUCCESS in 3m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/30b597394f9740c7bb2da4e007ccbee6 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ac1af76c837438bb84f6cd0551117df : SUCCESS in 6m 16s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b5d968d2306e4394ad36f12174294dc2 : SUCCESS in 16m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f2a65db1b2b04fc299aea55487db44b9 : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a3b29d0cc794189bb0ca6e00adb66a6 : SUCCESS in 1h 12m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c2a6d23716a40ffb0bb7d19fb6b150d : SUCCESS in 1h 37m 47s","accounts_in_message":[],"_revision_number":20},{"id":"742ae1862b98d2d032187334fd36e48024377115","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 14:10:25.000000000","message":"Patch Set 20: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2591a86838f1e4c0620305c2f083e1f9cdc8fbc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:57:57.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dc9d08f7ecd44f99a56cb6f95aa585e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eae41411da54835b176481faf126527 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505c161ed1344f38bf3701fc95456c92 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8490ae603b4a4f97a2f137de5edf4418 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/366929e0f6f4482182b6e7d79e2e3b74 : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e73a3c087ce5492da61249784c6e1328 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdb7ee4a73af4c078fb0ccefc8ffa54d : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a289ebcffc6c44bc9e4f990dc3a947c9 : SUCCESS in 3m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a117b422bc914ba6a165b52a4c495db6 : FAILURE in 3m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f6fd3ac329f045559fb5146fbb655b02 : FAILURE in 4m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dca2c8edf8a4471e9c1c33267021fe5c : FAILURE in 3m 49s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/370d002e00e44b43b19091aadb051343 : SUCCESS in 4m 16s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/bf1975734efe4c6eb7afbeeb5c764eaf : SUCCESS in 5m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/dfe83c7d997a4a9fa725bf5819c262ca : SUCCESS in 5m 05s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ce439ffd15d346e5b74ecd8b2693ff97 : SUCCESS in 9m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4bc2826797ff4060940997293f88e9de : SUCCESS in 3m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b5b90ca773174113a487414a5bb1543d : SUCCESS in 5m 08s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/277253e900154b0eacf3257d249b3d4f : SUCCESS in 4m 38s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1c5870d419b345f680b226c3c47c5dd0 : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0629526af204dde82d94f1903eb0baf : SUCCESS in 7m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/ca663e7ed406411d91461afad4c09898 : SUCCESS in 7m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0f9f3e2883b744d9b5e45a7dbccabe46 : FAILURE in 3m 35s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/6a3e088a297b462ca02a5fcc2c68679e : SUCCESS in 3m 20s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9dcaa0c6726842049481305e23ca00db : SUCCESS in 6m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/1bd23385810a4cc981f4300a70b25397 : SUCCESS in 5m 15s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/41b52acd42104fa289829ea81f19e082 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7b6d38083c8e47d8bf6b7ac484e3cfd0 : SUCCESS in 9m 40s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6a3c003fd1b84e0d8f5ca1d963334a4d : SUCCESS in 25m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/70e732bc05ab40bdace5700ea1f8afe1 : SUCCESS in 5m 49s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/bd27ca717cce47519129b140914b90e6 : SUCCESS in 17m 02s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a4881f6c8f8f4b998015abe3c9c3b347 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9004a886438341f696a654ea64dffc2a : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/364642165243443e9220a5a9f4b60059 : SUCCESS in 27m 09s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/f980fbc8bce3411c8854ca69b96ef5b6 : SUCCESS in 5m 55s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/df4e77a6e9e24a449f3dffe01862f603 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7489cdfd44ef40f5907c08c2bddcc2c6 : SUCCESS in 7m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2c6eef766aba4c8099566ff634cc502f : SUCCESS in 17m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3d3d820bb1974d70b13ef259c13083b6 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a6a8798e5e440d58935a4b8bd2a4763 : SUCCESS in 1h 08m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92e65572bb184be59eb77f8cef0be36c : SUCCESS in 1h 43m 38s","accounts_in_message":[],"_revision_number":20},{"id":"9e7dae7fd3da3d167accb29de14ea05f01ae5d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-21 11:20:18.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"2aac139ecb2924ce3c4f17293accc762aa77b2bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 13:01:49.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f2e29082f7449d1aa411ba7d5611929\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55af4e77affe4c4dabdfb0af0b10fcb1 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024a621d7d95470d90368553aa36c2e8 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f4d37e3a96a48939b53458107ae0329 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f359c5624f904e698516aa866b339940 : SUCCESS in 5m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95501fbbb95342db9baa5d00733430a5 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6147fe0cd2948a39c8dc350a0981fa5 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/448110f696f94e96a14a9772bb5ad425 : SUCCESS in 2m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45c9c79260ba4223b4792591a6614f2c : SUCCESS in 7m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a99925f2955460cb5da211021b21bde : SUCCESS in 10m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b7c0291a07054f43b8affb06ea8ef62b : SUCCESS in 5m 32s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b66e87c0b0b24f61a078c0a997af451f : SUCCESS in 4m 33s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b7cee7fa1852431e8ae73a827c4bf959 : SUCCESS in 3m 54s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/0c512fb65fad44c7bd7b5282fd9404aa : SUCCESS in 5m 00s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/3077a5797cb94c89a02d7103ff7edb0b : SUCCESS in 9m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2aa288a6f4d14fdab4e31e13490a18c3 : SUCCESS in 4m 36s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6c9830d8d91546128aa6c7f800e544f7 : SUCCESS in 4m 06s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/11a24ca9ce8644f3a4466c91b8bc07f2 : SUCCESS in 4m 58s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/461a287fe8bf437f8d4538c872d48670 : SUCCESS in 16m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68eeacb8ecc24f07b2f6e1f7cd3599cc : SUCCESS in 6m 57s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/ae467a7df71c45d096cc12f1c92a463a : SUCCESS in 7m 17s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/1addb6c3fa6f4f998b887af887ad8665 : SUCCESS in 12m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/dbdf1b18faad4a7190569e9f7f517b4c : SUCCESS in 3m 07s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0c395c4e4fa248af8ef00096cdd5b330 : SUCCESS in 8m 29s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/cdb402dea0af480a933fc5730ce97a4d : SUCCESS in 3m 48s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/765f30695e9b4bbda40a9e774deeee04 : SUCCESS in 3m 38s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/c9b80fe956fb4633847f944b8260404b : SUCCESS in 11m 01s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/18d0c37dd6e24ff5ae3a5be7c1cccc76 : SUCCESS in 15m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc236d0c3117468fac24e3502ecdca0a : SUCCESS in 5m 43s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/92e6dc3deb1e474982c48bbb52601d2e : SUCCESS in 9m 17s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8ca7c5082e5e414885521382f59a75df : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49443e34bec347e3846a9995cb0825ee : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c525bf7fd374487f8f90f193e16297a7 : SUCCESS in 25m 24s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/19514721a7df49fa9bfb0ae956e49101 : SUCCESS in 5m 41s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ca3695334f9141b2a97f7bb1b6313bdb : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c228112642594242b1e46c7c8b073ed2 : SUCCESS in 6m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a00391fe087146c38b24ba99e50ca047 : SUCCESS in 16m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1472bbaf39e945ad9387c6d3a42e75cf : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8f87208cb8c42d6959f71505a455f8a : SUCCESS in 1h 19m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4044cf1ee84f49158d73294d39b40307 : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":21},{"id":"86fb5bfb467ca27ee892537e9c60b19e9c7ca25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-31 17:45:17.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"c9770dc96c232818dcdab3360b3ebb8f394d79f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 19:49:03.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da320d9315441bab8fd2b61e112cbbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd54e5092cd84c4aa020815c7d9b06a2 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f3309712a840888e1ab9e2f5a5c61b : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/849c5d1947c846368e0692f681b0c560 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d832211d653e4f52a026ed148eaaf07e : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15a85bac235e44d3aad4c907febbdd69 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cd0f26759e747f5b3e55986fae55776 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02fcdf1526ff4687822206ac55ee4d02 : SUCCESS in 2m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/22d14fe8a25148eaa0ff3c7eab76f455 : SUCCESS in 5m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/500269f890494038ac1a36e33fb0d237 : SUCCESS in 3m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/81d81e60ae1f41b189293223f0cf95f4 : SUCCESS in 3m 14s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/2c4688bd9a1c4e50a720f1b8473a230a : SUCCESS in 4m 04s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/65c7875c5a6a4109a54b49007e37a968 : SUCCESS in 4m 42s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e5c47676492c4478bda21e77bc1d8e3a : SUCCESS in 2m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/44bc2fbe33f94923b671e15e15055289 : SUCCESS in 9m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/811a138066b542108b5a98d9cbc6a9fd : SUCCESS in 4m 27s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e34c0a3396cc4da99728d8e298775f81 : SUCCESS in 3m 30s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/4d01efd78aef4277ba75a78a6623f39b : SUCCESS in 6m 02s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7c4f7e4218f44d3aa7f1db4a0c1a1c8d : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13930afbb2584648b9aa0f548fec90a6 : SUCCESS in 3m 48s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/826bdaa3c38b49e692d84e71b70a0f3e : SUCCESS in 8m 24s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0b0b3a48a611417eb18a1d2795f6cdf6 : SUCCESS in 8m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a1f335fdfde0479faac118ef7dbc1076 : SUCCESS in 3m 34s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e49021b34a3a402a9fa8b7e55ffeb82c : SUCCESS in 6m 36s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/eb4b9efd59f245629a615f98d07a9ef7 : SUCCESS in 4m 11s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d29e6a24bc9b42af9f70a299fda95524 : SUCCESS in 2m 36s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/1190d665507545f8aad778e73fdb562f : SUCCESS in 8m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/008c0ba71b4c4610bdc6886022f802bf : SUCCESS in 15m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dd772fa5b79a40089ac526e40967d9c8 : SUCCESS in 3m 54s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d23366ef967040b3a0f5b73442d445b2 : SUCCESS in 13m 31s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/b140491fd8af4b038ae51b3bb77b970d : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f763cfed41d4a48ae7383ce42963959 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a48879814734c58bd5dfb33b77fe43a : SUCCESS in 25m 43s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2546b8db82f744bbb7417357dff0d090 : SUCCESS in 5m 04s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/d7f13457610b4ce99b02660c18945b41 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3aa5ba31d23e438cbbfeb06ce7c9547f : SUCCESS in 6m 07s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d6329ec109664ab2a05149b67b358552 : SUCCESS in 17m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/60e3f10909fa41ef8dc089bc038209b7 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2db1da8117614b2cb620c853134f3093 : SUCCESS in 37m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e03bf4d6a3f94736bc2b0d61a08efdf9 : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":22},{"id":"b93c5c148132baaddd85bcbba2af25f7d130590e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-06 10:45:51.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased. Commit message was updated.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"e141388b642705579e6d5cf200604f9451c56cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 12:48:06.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9756678b110d4a8684bb1b78b1933e57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9b13ebd5fe4b39b63812910cee2639 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb9355287944010bbd3c7991425e2ad : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18f6a9dd25d440c6841128f7087fdf26 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b718cec955784a31b201323461176e18 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fb4f2ab67a14cdfb22e7100af5c2f0d : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9424e172e3fa4dfcbe5f73c1585699ae : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5654b8a1670946388e5daff96c5907cb : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5bbf973a4fd9436fa38fefc1c08fbb76 : SUCCESS in 10m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cd400173dbd44c66b761193c9fd6ed0f : SUCCESS in 5m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cd6a67e6a6a44a22a7a6df29bffdf631 : SUCCESS in 3m 11s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e27a5819b44442079940a5a1f4b79ef8 : SUCCESS in 4m 12s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b0bc6c5611fd40f98044ce2ee89e4b58 : SUCCESS in 3m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/668394f5965b457683cf21d33b0e5d34 : SUCCESS in 4m 42s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e7dcca2dadf2499a898204ea0fba0d55 : SUCCESS in 10m 32s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6d9d45b8ef16446c9890b4c77f05e874 : SUCCESS in 4m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ff2c654c44274e90bf73ff31321d8401 : SUCCESS in 5m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e3666aeb1fc54eabbf473b1bfbe3857f : SUCCESS in 4m 07s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/cef618823f494e4abcc34f338dc3055e : SUCCESS in 13m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/315f747dd8fd497a958b3a2c0c94b002 : SUCCESS in 7m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/51dbe428d97742aab201b3f1454ab0c3 : SUCCESS in 7m 13s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/79a431f491d04f7db6809a272e813eef : SUCCESS in 10m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9b469878971f4478a7669805626985f1 : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/a2347a6eafc5425aa01a4b676beb025b : SUCCESS in 7m 39s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/cc7dfe6e7a974a0da28091730cd5f209 : SUCCESS in 4m 14s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2008db55fafb44059d05b2c1c8fd5a4d : SUCCESS in 3m 13s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/edd864487b144d5db62fa631f47867d7 : SUCCESS in 10m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/bf2b18b862cc43b693107ba2fbc099f6 : SUCCESS in 24m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f540d1431d1c4b4296218e580641fc5c : SUCCESS in 6m 21s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f452558b194d42fda6d4889a37bd1727 : SUCCESS in 12m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c5f956d57c1942d989024d7d8e3769a8 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9dfaff8f1bb04b6a9772b10554d1042f : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f23858091e344e5aa879faacaafc0387 : SUCCESS in 31m 11s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/26bfade5120946eb9018cce4554a1a4d : SUCCESS in 5m 23s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bf443d113c9446359695986bce68c406 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13d5555441824c5a9a8822eb8a7f1fd7 : SUCCESS in 6m 32s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c7a8cbb9db8947ecaffdb5d41267fe1c : SUCCESS in 18m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3846b3734c364f8b97cb23c34a57806a : FAILURE in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/658e9185cda440eabba02fb6fc8e7e90 : SUCCESS in 1h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b587aa68c5f0412483cbfaee77e3c2ed : SUCCESS in 1h 55m 53s","accounts_in_message":[],"_revision_number":23},{"id":"ab5f8598431b7beedf054911c64c2c4a08bf8781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-24 09:56:52.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"e48977ec04214cacdd254798316707fbc5fd9778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:46:39.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b450f6e2eebe4ceab3704c9dcbb32107\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f860ac81c4ac49788e427a897e196512 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f35c513ac1644d6915791b317e69d08 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/954deb24691d49c499488dc748a6dcbd : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35559464acb34bb990f45611e9d44544 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45b8c613009e4d2a8cb2a70ffe83ebd2 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee476ddb78ba4959af0559d94c49e01c : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9228c03a7cd04050b8f80573f5dd7c90 : SUCCESS in 3m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ef5e402f6e344d26a1ce18228b5c4473 : SUCCESS in 5m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a996418682824fefa67b62c9fae4f101 : SUCCESS in 3m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8443db7bc0a64d348da972a601ce195a : SUCCESS in 5m 08s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/952f427d8335490f9bdf7044492feb26 : SUCCESS in 4m 39s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9c68eee0aaab422cab86d3e2c8e0375e : SUCCESS in 4m 43s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b5470b4312c94b9886420e78baac3717 : SUCCESS in 4m 51s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/328c9416b95f4485a231de7c4e2d7df0 : SUCCESS in 10m 29s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/21cf186073d147d3a8e4513c708fc3a5 : SUCCESS in 3m 32s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/8b7385c1bd6d4273bf3caf94af2ada93 : SUCCESS in 4m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/6e61ee7422834343b6d5cc22ff8d2097 : SUCCESS in 4m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8334fc46070b429cb92a0e492d7b32d9 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5886d9d9298c4a03a9226f6d01d3ee96 : SUCCESS in 6m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/91f695a205d84dc39c2dea21362a96de : SUCCESS in 5m 11s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/65fff49bf39c45a5a79a19d4f8faf360 : SUCCESS in 11m 13s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e714324daa864f53ad768cb63b39f436 : SUCCESS in 2m 46s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9e38cb6e8d864ad78e20f5847290c912 : SUCCESS in 7m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/8d67b10075db49a584890e6a339ff0de : SUCCESS in 4m 32s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/833083753f0d4a139b133bc9c03666c2 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d94ddc65bd3a4e2ba80c8f6afa74eb7f : SUCCESS in 10m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/3b9e40ed32ac4cfea347dc62123fb212 : SUCCESS in 26m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/055abd5f3828426890aa5a2ae810ddf2 : SUCCESS in 7m 16s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/3f0314efc274430a870ed666596830bd : SUCCESS in 13m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c2e3406908c44de1815e5330bd50e73d : SUCCESS in 3m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f54771544f247218694e5c9a351918b : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b874301d765b4121b0f0057ab3ffdc3a : SUCCESS in 24m 56s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/9f5a0dcf67d14391912e715ba59723fb : SUCCESS in 5m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/1e1ab036989d44e8b6082d64e4138bf4 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39d47b069a794482b0ebe9f5a5e20a8c : SUCCESS in 6m 09s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6617f5ee16d14101843f32d6d69ae13e : SUCCESS in 18m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5439f94bc7484c4eba2610483efd952b : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4de072f3c3aa4066b70948634fc25dac : SUCCESS in 1h 07m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53f36b80f13543c5a597896e8a793f20 : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":24},{"id":"b46046b622b222269b75f6017c34d2052021f269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-28 10:11:35.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"256d0f45f87ef959fc24a97dabe81b222630c71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 12:05:34.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ed267e48d74a70bde14d20f7cd4dfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f06198fdebc4c48b80c55f5e4702739 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d01d4cec4d14bc1b41a9b4972da828d : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/631c22e146094a8b9a59be4ae47f9b95 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d00423ef143a4476872a41a4b68e5710 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a58078b21c9e478181bef2e4b820339a : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f269cb5f75f44808865d22e1b1988e3a : SUCCESS in 2m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7128250f61c4c97acd8f280d4f93a69 : SUCCESS in 1m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/280a628264a64d2589f1f7710f171dfc : SUCCESS in 5m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7d4bf3b0d1a4d539bb859f859cafad1 : SUCCESS in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cffbed926fa443ebbd690ec39508eef9 : SUCCESS in 4m 23s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/00aaf77b1df143cbb707bac4eac46939 : SUCCESS in 3m 57s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/966544faf76a4d26967cc8a31e5b6f81 : SUCCESS in 4m 28s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e8e10865121548d2accb3c25dbc6235d : SUCCESS in 4m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fa8a42455aee434eb15c1f3d46253320 : SUCCESS in 7m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d51b013afe38432e8574ce05841a27f4 : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/39d68f8a4cb54f50a91a3ed056609339 : SUCCESS in 3m 11s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1c9572cbb8f44c4f9c457b71c8f37dbb : SUCCESS in 3m 53s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/f2512c5de73c480699c9d269c82a168e : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d311268cf1874ad980ab7c0173467033 : SUCCESS in 9m 19s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/62d01760aee84912b3caffc0e467d251 : SUCCESS in 7m 12s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ef2e63811fa44fc29905403c3b69ec6e : SUCCESS in 8m 19s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/11215b15ace0412b9b2472ade3824429 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/3c88341312944c44beec1ea2224ecc22 : SUCCESS in 6m 29s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/17ab85d1ff2447ee89908363ba9e2a53 : SUCCESS in 2m 33s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f5ddf7c8423f44ccb94d05469ee099ef : SUCCESS in 2m 55s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7f36774bf2704a31a2b3faefb4af862c : SUCCESS in 8m 59s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dbe4205adfa2449ab4a988fcd90ffe6e : SUCCESS in 24m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ce59545fe9847928cd608d5948adf72 : SUCCESS in 5m 59s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4c91bcb35b4e4495a77212d4157147ac : SUCCESS in 9m 49s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4ae7611f739941e2950cbb744dc34427 : SUCCESS in 2m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6a3c03b709741578a0f43abcd4ed670 : SUCCESS in 2m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89508fca30274c86b5a365156f198825 : SUCCESS in 19m 39s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0aab1864353b4dcba407d103e75099ac : SUCCESS in 4m 29s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fdd7eaa05f2e4cc28de16a860a5b748e : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1830756208e04a93b0ebadbabc3e48a6 : SUCCESS in 6m 09s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d4fa7f23c96c49a0a95ba58263c474ac : SUCCESS in 7m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a20849ffd8b94b9995a18903fd83ac12 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/770def5b04e24d929e3fa57ae3ff5601 : SUCCESS in 34m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf49343dd18a435e90e57cf71498fa54 : SUCCESS in 1h 47m 39s","accounts_in_message":[],"_revision_number":25},{"id":"1bb48cd2f0d38765aa99edcb340442dd75eceffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-10 09:59:25.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"e0dd1452de00d58a386ddfeb24dcff45eda116d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 11:52:29.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef89b1342821424d9079fc1cb1eb907d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c3268e5c41440283539627ccb0c7d3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9baaf9737e343a2a3a6eb1b7049d957 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/600bd71b5cc441ea9ebba0f690ef60e4 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9557e637c6e4f56b4057c38b2534e7b : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65ec14a42dd84c6e982a3c4dac9caf4b : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce21215c04514c96813975f714b58618 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec773350ccd745729062b696cdb866d2 : SUCCESS in 2m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f891f5be3dc24f02b1297799efb6ef1f : SUCCESS in 5m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a6e7e1b3db0c439faa06a6567f0dd683 : SUCCESS in 6m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d669845292f54920938334d333cb8eef : SUCCESS in 5m 24s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6a111f222137422e87818ddad9498965 : SUCCESS in 3m 53s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/81355264a007417d924677f28cbeb59e : SUCCESS in 4m 36s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f0d2bac77ce4402891664bbed8e10814 : SUCCESS in 5m 16s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/19d94c323001443698a3c864e49e236f : FAILURE in 10m 24s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/931d45867e534621a0520112cdc3ee6b : SUCCESS in 4m 35s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/79cd878dec1249e78451e5da68b82f01 : SUCCESS in 6m 18s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/99ad5edbe5f14350ae4617e6a92ed0eb : SUCCESS in 5m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3d41cd28cdb04b06ac9479ade10a16e4 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7081f112f70a48faa4a364fb406efd3e : SUCCESS in 6m 49s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/d1139db6522f431c9e864079087f2051 : SUCCESS in 7m 18s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/566af4ae479d495ebf8087180c9fe956 : SUCCESS in 10m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/10d906ef2b3d486296eab81631212947 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/245bdc9038c047669eafa95846a5dcd6 : SUCCESS in 10m 21s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/9f98811011cc40eb82c0af75aed5e7cb : SUCCESS in 4m 02s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0c8c46fcab904d9991eaa6a495d59013 : SUCCESS in 4m 16s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5f5622c72df342afa74c3d0f8033882f : SUCCESS in 11m 05s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/070df85b435748398b688a3466dd0a33 : SUCCESS in 24m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/88f23228518845d0bd5cdf1dbbf586ac : SUCCESS in 6m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ec4095d6d47e47a2ae46027e282f21c4 : SUCCESS in 11m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/04c70adb97344f719e160dacbd62cf64 : SUCCESS in 3m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f5f2491e4884fa4802c3d347c0dbce8 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7de989ed75824d18bd6fd12e9d730d83 : SUCCESS in 21m 26s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/bffc18da04084ee2b250c2b6cd1efdfd : SUCCESS in 6m 05s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5a99e910a1dc4e988e2418745dc350bb : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/058e2d4d804b4a4a932212ca9b4fb291 : SUCCESS in 6m 24s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9e50a9c5be934e5b85896aa38af40343 : SUCCESS in 16m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dbb6639d704b4fefaa20c96acd2dc72b : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d992a0872337497dbd27d4bdb7d8b534 : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/245f0d08b73f4d6e93962aa2a592c42f : SUCCESS in 1h 45m 40s","accounts_in_message":[],"_revision_number":26},{"id":"ff15adbbac6b4cf247727bb88a2ef64a08c1e997","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-11 16:22:34.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"5987c6a627879f184549d8f5064b529dd6c6292f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 17:27:01.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae36efaadb449e1aa3a4d10efbc8c1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b842e2fd274ee48403c3be4e6fd808 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1dfd2450f54982be2da5140e8e2ea2 : SUCCESS in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0bce7433069045aaa245ec68fda675ec : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/951c86cffa0b4beab3d0f7d280e9d801 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae5750c747f941efbcd130afcb181c8f : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a152e1a67904c1fac55b80afa3cffb7 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8301d89882674bb18a7c3ddaff0e3554 : SUCCESS in 3m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/48e02a10b3e24990b30b87ccb82a73bd : SUCCESS in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/955f36b8043d456ba505e0b17c390dac : SUCCESS in 6m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c52cafaf71f8441d8c20c141f2891c5b : SUCCESS in 5m 25s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5d1ae48e71c44988937d4ed942e8ccc9 : SUCCESS in 5m 57s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/41b1e678be2649298703a93d4a4b50f0 : SUCCESS in 5m 20s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/6e4ff28aad094f88b6ce15e205cbbbed : SUCCESS in 4m 08s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/bcb7ac7b796d417cb83d70d0305b9caf : TIMED_OUT in 40m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f353a34cf6eb4861bb4a298b019ce4ea : SUCCESS in 3m 21s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/21bb09f893bf49abb55bd346947d99cc : SUCCESS in 6m 15s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/eab0e11ae12f4114bb281b806705c9fa : SUCCESS in 5m 31s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/dafd2fcdd07d4d6299fe7a54325d0957 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34b8fac25a7b4a7183cf368a47d08343 : SUCCESS in 11m 23s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/32feba066cd64fcd9aa77bb314636b57 : SUCCESS in 7m 24s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c63e67986bae48a4bfea828989028d77 : SUCCESS in 6m 51s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2c46b0907a8d4377a612f6336024a81b : SUCCESS in 2m 01s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/483e5ac6fff347d3b3878451cbce066a : SUCCESS in 7m 40s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4b6d445eb6714d0ab44edba1cb212676 : SUCCESS in 4m 24s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c4c7e175a4c24108abf28e63e30192cd : SUCCESS in 2m 38s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/44b7e40629114d06b7536ec9a2945e99 : SUCCESS in 9m 44s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a02afce443154f1e8d0d6347fbf32e39 : SUCCESS in 24m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3dac05daa8944f9099234e4c4e5616bf : SUCCESS in 3m 30s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/3eb0d909ff60405aa5a12af4ecc85ed2 : SUCCESS in 12m 57s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3cbb9dfb248b47fa954139ed96b7ec6b : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/513a9e8c912e4089a6c91896769d1bb7 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa26a07f1f794372ab4352f306cdeff5 : SUCCESS in 21m 00s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/0389fd9d571a43258617b6b35f2f0cdb : SUCCESS in 5m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e88217ce84e7454ab2224d6119b3700f : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28d6799369f640f4897271a78ced91a0 : SUCCESS in 4m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/80b1a20d5f1d42f7b3e114db0e8fcb79 : SUCCESS in 19m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ba46834a7ff432c95c3a0e7b048660a : SUCCESS in 3m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a92d7a7100f14d2a97eb708d753b8788 : SUCCESS in 32m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f61095617c2046e0bae3161261fe2ba4 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":26},{"id":"513d071413d9b6250fe026d265d30a5a6b69bcd7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-11 18:00:06.000000000","message":"Patch Set 26: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"53885730975685da09925a4a515c91accc1a464c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 18:43:25.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0304d38374094ed4ba477b0a91f00b02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03469c9482a84582b27f4d6cdecea53c : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f282b9f182e42deaad645c325cea218 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5223d9291caf43af91613f5b725cdf41 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/709aa0b3be6d4b94879e4ca26acccbc4 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f6ba366453a0463291bf55b061d2e057 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a73d1d52e4945958cb72a46a4ffa4b4 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7161063758c94867a62fb6dd8ac72c4f : SUCCESS in 2m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/22a138062ffb4009b0c45c11f3f24b8e : SUCCESS in 5m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/48d3ff95b2dc4703af1550f9cf30eade : SUCCESS in 6m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cce78680674143b3921a1a8c9e8fd840 : SUCCESS in 7m 13s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e288aea633484c0f94767829bbf8c7fb : SUCCESS in 4m 09s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/c33b6818d4ec4fb0b808a367f082601e : SUCCESS in 4m 25s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/b0d20568fdfd4bd88077cefdd6563570 : SUCCESS in 5m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/eb379359793d496888f34152e6adce01 : FAILURE in 8m 07s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2559552b29ff4b3d985897616822d770 : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/eba795e51c60437f9b7da9b77713c67e : SUCCESS in 4m 50s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/caefa50474054766bbef4c59ce538fdd : SUCCESS in 7m 18s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/fe8928e0d0414ee29174019d48230e92 : SUCCESS in 18m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/968aa3c49f1d4541b39646a1de298a56 : SUCCESS in 9m 06s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/28234349fa294e1db03e91a67517a623 : SUCCESS in 7m 09s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/08e4ef7756cf4188aa8edfe54dcee612 : SUCCESS in 8m 45s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/971a6215973d4994afbd8c9121b09076 : SUCCESS in 2m 13s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/113f858fb364401da771475744751b43 : SUCCESS in 7m 04s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/76bff8ba4b344cc9ab23e78d61fe982f : SUCCESS in 3m 50s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/cbde2d55671341878485cb24951171a8 : SUCCESS in 3m 51s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/dac06d7f26f94532add32e91c8c7d294 : SUCCESS in 9m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/5dc0d94c432c48218d1f4bb2988dbb75 : SUCCESS in 23m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d9c1c177884d47b98febb4775e0b6df4 : SUCCESS in 6m 14s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/85bb02666aaa43e8a74a8720740498ba : SUCCESS in 13m 48s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/653c84ae970f456b8e28d8c501601859 : SUCCESS in 3m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52dcf426cc5b4677b3c5a2044c223302 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18f7f2d06ff2436f944dcddcbb8747ee : SUCCESS in 24m 34s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4545a002d9874f03bcbbd933ced9ad08 : SUCCESS in 5m 22s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/60b3b0df5e9d4eecacf5983045fc2e37 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/297eb7c7eea649a28f008e9375242927 : SUCCESS in 7m 36s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d962073ce0d746a786634659473d9799 : SUCCESS in 9m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/af6e4ce363024fa59b0a5be7b21f77db : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfff02945c0b4b67be920317ac20a5ef : SUCCESS in 33m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a033575a0528497a8c14ca59ce8d0e18 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":26},{"id":"877cf04664195a6a3a6350d92ddafd6781d3d88d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-15 09:00:30.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"3be63bebb5af273b727f58d818f36bb0fc193b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 09:51:03.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cc1e47f1aaa47e3b40e8b461571a7cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f9b9aa62e543f3b592a2e0516d6574 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba1466227ed429ebce6e748556e0d8e : SUCCESS in 2m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6afcfc91eae242a4b76b727e503ba8d8 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0b0d7adb7f549fcab100d8a2cbea73d : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c73144435d244bbeaddcf4da4e0b64f4 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4e0a7887dd745009a1cfc48bc182c8a : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e377770eddf54726962b3710704809b4 : SUCCESS in 3m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/561b082ceb1847689fa45dd24d388593 : SUCCESS in 4m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4b13e4012c5444dca681f89c64f6eb98 : SUCCESS in 3m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6ddbd49b47eb4b0bb5b681c04385da69 : SUCCESS in 4m 07s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/9919fe4d21be472e858f78f3d972c9df : SUCCESS in 2m 46s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/cbb991e65ef04ee1b3a41255b47558dc : SUCCESS in 3m 05s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/faca55a11cd441a49ff361739e6a5026 : SUCCESS in 3m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/839538a4c68a4a9d9dc1b982faff1334 : TIMED_OUT in 40m 59s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/265a57bfa11443e58477cd6db4635e49 : SUCCESS in 3m 17s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2e145e23221d48ccb717b91d47d08cdb : SUCCESS in 2m 55s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/51a068763b6a4c66baace34123cabdcb : SUCCESS in 3m 44s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/ab5155e44a1b48e784607dd10b919c7c : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de9507a195be4a738197b905e8a1f8ca : SUCCESS in 4m 09s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/080d657fba1542f997e8d64ccb5db55c : SUCCESS in 4m 17s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/99d6b71ccd2a494baae2374c2b6ad873 : SUCCESS in 11m 17s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/20cb33c1d3664151bc8ba920199870c8 : SUCCESS in 1m 44s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/5ef9d9b11e6d443fad546d07792bb28c : SUCCESS in 4m 37s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/11086002bd134092951ed10997d72e9e : SUCCESS in 3m 35s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/bc650a56cd8242f5beb54542deb36ef5 : SUCCESS in 4m 04s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5aa9f70b28354648889a52eebe846b85 : SUCCESS in 10m 11s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c8b3ef50c8b54fada91920da0613ba7f : SUCCESS in 16m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e13e11793f2b4114b8f47a67066acfb6 : SUCCESS in 6m 40s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/21720617f1b3479f97821f659fa5d4e5 : SUCCESS in 15m 38s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d49698c1c5724598b8d7423426c7c29b : SUCCESS in 3m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/12f0da81ec7148159d4dbabc0644647f : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16b7b9ee0c2c48ee95ed2023b2683275 : SUCCESS in 24m 24s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d07409d7d36947a4a1db36ea9e4ea55b : SUCCESS in 4m 36s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4238c9ed1e684b84af9937ed5df1d0e5 : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09d9a015b1fc4fa890da24072c72731f : SUCCESS in 7m 00s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/3e7100d5263f4510ba57db3f59f597c0 : SUCCESS in 9m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/684dfe0888314991a74758f1b37049ae : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b3679ed57f242089b2cd7e12f92dab4 : SUCCESS in 41m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2614d0b36adc45c0b1c10668ecfeb081 : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":27},{"id":"3e4bc3f9adeefdb0028c9301ce2694221b2ea653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-15 12:48:52.000000000","message":"Patch Set 28: Commit message was updated.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"385f8b450633727cd05db3731ec15bd2a9c2d581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 14:32:49.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d723cb87f7e49d2bb87e90cdec92272\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d832c23561844a9daa95935681839498 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c516efadfa4da894be1390587b634c : SUCCESS in 1m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abaf46dc24664429a08ed6eff2f9bd4b : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fb969257eb74cb4aa4c7c33a723c196 : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a15b4fd833e4310b77805ae3745b56c : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d891dfff7d5453e984cc57cceb13115 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bab5549313884033913daebe2d36b94b : SUCCESS in 3m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dc8e5ba6fc204ec28cfe1220ed91eaaa : SUCCESS in 5m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d594d1c0f6144b418889119d12447d03 : SUCCESS in 3m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67182faee2e4401eaf11ac40385290a4 : SUCCESS in 4m 54s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b365ad6694ed439e9e68b27533edc409 : SUCCESS in 4m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/51c502ad89b44f7bbaf79de372b5b8b4 : SUCCESS in 4m 34s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/276363c7f04a473496b700bd7dc4cdc4 : SUCCESS in 3m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e8c764af2e804280bd3ff9140486c1d7 : SUCCESS in 10m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5d58db84fa8b4291bdd398dab7ed6f3d : SUCCESS in 4m 42s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/2f285249ca104b20b4c3ee0c5ab4500e : SUCCESS in 4m 34s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/d2e1400f02ea43e0bd878ac063915e28 : SUCCESS in 6m 52s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/77ed607000074ac9af0e4b3e76c62e40 : SUCCESS in 17m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18e42b9e0ffa4dea932426244673ee5a : SUCCESS in 4m 51s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9ed10df336df48e1b9ccc9285db13747 : SUCCESS in 7m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2e69e3c8bad74ec4b65f0df91327b009 : SUCCESS in 10m 52s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d5fe960af5204f80ae9a52e41b3f6ca0 : SUCCESS in 2m 10s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/302e7bac61114873bf3e0adffdc831de : SUCCESS in 7m 00s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/abe4ea21058e45eb99a5f08b196af466 : SUCCESS in 6m 57s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0a96ba92f3a440c48bce541f62bc0f14 : SUCCESS in 3m 11s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b73bb421404a4b31968daf243f0a84c7 : SUCCESS in 10m 26s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0eaac73bf3174769a80307a724ec9e6a : SUCCESS in 24m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/19825c3e45e14d9e8687336f9138ed50 : SUCCESS in 4m 07s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7b2d3753e2c3479a839120cc55227d45 : SUCCESS in 12m 59s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/872497c9f14a48a8bd1dd8a2248a7a67 : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0716a13355594292b063dbcaf6014576 : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/695933695cc743c495cb0a8a21269eb2 : SUCCESS in 26m 44s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ad8d9925d49f4d4780a74ca725b85cc5 : SUCCESS in 3m 38s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/d04653f13684491f8c7050059024fc6a : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd0bf53075f4492d97dac88818144ca4 : SUCCESS in 6m 43s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/fc5ee7d4eb4c4ba9905f0e680d39dc37 : SUCCESS in 15m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3da72b77ddba4b3ea31c995e6bf81277 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/559a04b24c35463dbf2cedbc6f3b4d7f : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d971f40ce93341fcb710f8ba9c878cef : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":28},{"id":"754dcd188ec97f484108dc37940c6029d2779d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-05 12:08:55.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"d52dc4bf3546a8fa8ac5922e961b4179be7b8d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:30:50.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dcf834fcde84ebaaeb383c9124f0090\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7b79cd712a4f8195030a6852ea1562 : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58459d6238e4b18911946466678671d : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c140c635c14410d915988ad725d74f1 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa635376a59d4f90a3a2227631130328 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b03da83757b4ae89573e0c1af7859e9 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c6c67c26b0d444292d5e8c9f04b4b80 : SUCCESS in 2m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b2c2830c6d4d4fcc9b09a1594751689a : SUCCESS in 3m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/11ec8eea4df74537bfac49f0f0929f1b : SUCCESS in 3m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ed8b07c03e8d436499bd3806a0f78852 : SUCCESS in 7m 24s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/aa4b36ef4efd4e3f8252f2212d006141 : SUCCESS in 13m 35s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/78f9be07a5334fb8b6285d3db2548c1e : SUCCESS in 4m 55s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ccc29581cbd64e44b3f77f16702da15f : SUCCESS in 4m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/de51240e319f4a20a7ce556b1509483c : SUCCESS in 3m 08s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ca6ffd7712344c059d17eaf1d40c38cf : SUCCESS in 10m 27s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/21356856678a4bc6af34e975a316aacb : SUCCESS in 5m 15s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6766fafa023b4a22a11a6e2ea00030d1 : SUCCESS in 4m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/59c155c6ce2143849537e804e3b72b0f : SUCCESS in 5m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/17c4235c6c674601ba8d17c652fae586 : SUCCESS in 18m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bb1d563702a4627b5487da992035abb : SUCCESS in 6m 20s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/884e862033dd4996bd52fb537c1ac2b3 : SUCCESS in 4m 18s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/c112f3be93ae4c009168c22eb1e4b2b1 : SUCCESS in 7m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/2f8bca40f37f42888b25988ed4a930ef : SUCCESS in 2m 09s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/36d9bc441fd84fb9bdf6de03d098e68c : SUCCESS in 7m 20s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/98de23407431454e879e1bf1869a5df6 : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2af19920da024459a4f1d5e31c53de52 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/7126449bf4a84b528ed0d599d06239a4 : SUCCESS in 10m 04s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fa7e514fab224962a77b9813ff645715 : SUCCESS in 25m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a7bd84ae21ce4332ac18ebdc43833ec1 : SUCCESS in 7m 02s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/58aa288a02f9461fa42a5cf1efd8002a : SUCCESS in 14m 04s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2bc89195421743df9fe2a30ba7036c80 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/209294c02a774a69b0eb576778ad7344 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/293615056cc142bf9468b59075c30eb2 : SUCCESS in 31m 43s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/f7f5c0814fa94741b87cced9f2a48cde : SUCCESS in 2m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/7d582800dd0e4e4ab422a3359d42b4f5 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6a358d87f44466c85c92266edb84eeb : SUCCESS in 6m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6c54c1a7f2e54bbe8d9d4d53f7b0de79 : SUCCESS in 15m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/89ab2a7a42074aaaa3e03e795f77f5c6 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86afaa493a5546f8b18ead45374ee0f5 : SUCCESS in 43m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbfc7398c5c2486496adaaf4d5526dd4 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":29},{"id":"3eff17def638fd25729359b1b7ba5ffc27a6ae46","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-09 08:47:37.000000000","message":"Patch Set 29: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"8e4db059ad9c1ba6b529c3a24318a7ec04ab0cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-10 10:11:18.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"8d527350f60d43774561c9a4df1947419ced554c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-10 11:04:11.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ccb857f43934648a42e0ba690a4e32c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e899471985448e97db88963edcff40 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151325109889405781fd7269a58d96f8 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34ff41b134a94014af7952a9c4937b85 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4be4d2f719241ca9f4582f0914a3fc6 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/632ba34298564c6788ebbdd5b5c51bf9 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/007fc0f74fe14940a41ce9cc1937761f : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2a7d2b8c57a84a82a17d9713604214be : SUCCESS in 8m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d572d41d7f934aebb2b84f73a9668363 : SUCCESS in 8m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0e5e032fb8c24992867da8899bd5f6bb : SUCCESS in 9m 41s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/e92c3af7610442aea3211470398abe4e : SUCCESS in 4m 51s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/53e5e8942f984fffb49eb36cd84e74c5 : SUCCESS in 4m 35s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9ceaf8f6324143dc9efb09a845ee73f6 : SUCCESS in 3m 16s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5b71e734e64c4d7da377b136dc4afdd6 : SUCCESS in 3m 37s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/5ad0bb2168d04bad942ba633c1c16c2e : SUCCESS in 7m 41s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/63c238b44b5f4c3da14bd256bab298fb : FAILURE in 5m 05s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b33cd87fa3e544d3a00e439a74dffece : SUCCESS in 5m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/eb096c36e0e74a0ea713d74ef6b798c8 : SUCCESS in 3m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/67a7dcd3e7c24967865efb2d341de6a5 : SUCCESS in 15m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4161a526e4164f47918422c06f0906e2 : SUCCESS in 4m 32s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/39eb9bdb956b4c6199c47a301de88c63 : SUCCESS in 6m 57s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/3ec44776fb3042a89fd485951ec287da : SUCCESS in 7m 24s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/1e0ff1c444d44811b12e48ad30e4d9e3 : SUCCESS in 3m 07s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f00084a136f54f46acb51443cedd2979 : SUCCESS in 9m 45s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/dea5b2748b734bdf845e6fc09a477f64 : SUCCESS in 3m 04s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/55e7a2721224404881492883e0119944 : SUCCESS in 7m 11s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/bff6d850786c4d12b1e08ceb89cc1ddc : SUCCESS in 9m 32s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8d501a41dddc49e3b658bd386501d770 : SUCCESS in 17m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e6dd351044f447ffa4deadd17b7e2ef9 : SUCCESS in 7m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fa9d8674935440b588fb4c0c19bf5f45 : SUCCESS in 12m 33s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/28af04af3cb34ca4af79db5bdf96f9e1 : SUCCESS in 2m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68d16fcf59234ba79cc676d53ed9c932 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34e446b6f9f942ffab110b1f01a70d69 : SUCCESS in 28m 05s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/81e9fd14e3644e2d91e8e29a2b1dbf2b : SUCCESS in 6m 11s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f3f66bc5b02845ad8571435c53fe5f09 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df56e14213f54f3483caa2c9d2eb30f1 : SUCCESS in 9m 02s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ac836db871bd4913993b041d9d5e7bc2 : SUCCESS in 19m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5619c78eabe84098aee4be24e16cb5dc : SUCCESS in 8m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8a3527ff0334ee8a55ad7c958b50400 : SUCCESS in 47m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b7ea146800e432fa1273cf83a590543 : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"702bd5d0906e786a7d0d80b43ee62d27d406aed9","revisions":{"f19d192824e77e527a74d7fbb602b576ba618dff":{"kind":"REWORK","_number":1,"created":"2026-02-21 17:18:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/1"}}},"commit":{"parents":[{"commit":"8b4ae52af8c192a964f83d8164170cd7b460301d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b4ae52af8c192a964f83d8164170cd7b460301d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-21 16:38:42.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f19d192824e77e527a74d7fbb602b576ba618dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f19d192824e77e527a74d7fbb602b576ba618dff"}]},"branch":"refs/heads/master"},"90700f475840c99f66633d6aa81dbb2dafa84582":{"kind":"REWORK","_number":2,"created":"2026-02-21 17:44:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/2"}}},"commit":{"parents":[{"commit":"8b4ae52af8c192a964f83d8164170cd7b460301d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b4ae52af8c192a964f83d8164170cd7b460301d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-21 17:44:07.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90700f475840c99f66633d6aa81dbb2dafa84582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90700f475840c99f66633d6aa81dbb2dafa84582"}]},"branch":"refs/heads/master"},"9f3869b172f675af6477b3dee8fa6923fa84351c":{"kind":"REWORK","_number":3,"created":"2026-02-22 07:50:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/3"}}},"commit":{"parents":[{"commit":"8b4ae52af8c192a964f83d8164170cd7b460301d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b4ae52af8c192a964f83d8164170cd7b460301d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-22 07:48:49.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- testtools:\n    tempest: \"AttributeError: \u0027AttachVolumeMultiAttachTest\u0027 object has no attribute \u0027skip\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f3869b172f675af6477b3dee8fa6923fa84351c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f3869b172f675af6477b3dee8fa6923fa84351c"}]},"branch":"refs/heads/master"},"a4c0a2ac647b16fdfb61a54af43d0127ba95d0c6":{"kind":"REWORK","_number":4,"created":"2026-03-07 09:45:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/4"}}},"commit":{"parents":[{"commit":"855c649f750e3dc0b69a26ffdb08284e09cf5d4f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/855c649f750e3dc0b69a26ffdb08284e09cf5d4f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-03-07 09:45:29.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4c0a2ac647b16fdfb61a54af43d0127ba95d0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4c0a2ac647b16fdfb61a54af43d0127ba95d0c6"}]},"branch":"refs/heads/master"},"80399969d4adc7137809b2da4f958799d9c89957":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-14 09:50:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/5"}}},"commit":{"parents":[{"commit":"82bc1dfdbf0ce5efbe09e54086d0f6c6cbef4955","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82bc1dfdbf0ce5efbe09e54086d0f6c6cbef4955"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-03-14 09:49:42.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80399969d4adc7137809b2da4f958799d9c89957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80399969d4adc7137809b2da4f958799d9c89957"}]},"branch":"refs/heads/master"},"b7f440c45fb449842a7813cfa00c5ddcb2dd19ab":{"kind":"REWORK","_number":6,"created":"2026-03-23 09:31:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/6"}}},"commit":{"parents":[{"commit":"34f4dd62a33632369d77dfb2d3d6786066c53a65","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34f4dd62a33632369d77dfb2d3d6786066c53a65"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-03-22 18:55:02.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7f440c45fb449842a7813cfa00c5ddcb2dd19ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7f440c45fb449842a7813cfa00c5ddcb2dd19ab"}]},"branch":"refs/heads/master"},"a8aaaa91ba834e35de7ee1c5b8724dccef4c9781":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-03-28 18:21:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/7"}}},"commit":{"parents":[{"commit":"aab3957b7b830da5f1cca934cb3043462ce76b9e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aab3957b7b830da5f1cca934cb3043462ce76b9e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-03-28 18:21:20.000000000","tz":60},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8aaaa91ba834e35de7ee1c5b8724dccef4c9781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8aaaa91ba834e35de7ee1c5b8724dccef4c9781"}]},"branch":"refs/heads/master"},"dd7da32352817c185851cd82aed639727e2ee56d":{"kind":"REWORK","_number":8,"created":"2026-04-05 09:48:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/8"}}},"commit":{"parents":[{"commit":"c2d957cbbb1b8fa10453063cf5075ffcbab80361","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2d957cbbb1b8fa10453063cf5075ffcbab80361"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-05 09:48:00.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd7da32352817c185851cd82aed639727e2ee56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd7da32352817c185851cd82aed639727e2ee56d"}]},"branch":"refs/heads/master"},"ff9c59a09d9e792cc91b7fd39b21e10b0b192d5f":{"kind":"REWORK","_number":9,"created":"2026-04-11 10:06:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/9"}}},"commit":{"parents":[{"commit":"3064274b8f7d6a985b841061b2b959f89f107163","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3064274b8f7d6a985b841061b2b959f89f107163"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-11 10:06:09.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff9c59a09d9e792cc91b7fd39b21e10b0b192d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff9c59a09d9e792cc91b7fd39b21e10b0b192d5f"}]},"branch":"refs/heads/master"},"93903298eb7a3cac1b0ec1bf5cbdc626cb3d1b8a":{"kind":"REWORK","_number":10,"created":"2026-04-11 12:55:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/10"}}},"commit":{"parents":[{"commit":"20e0b8ed4b383a6b3ca71a3c6942f21bb5bbe22d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20e0b8ed4b383a6b3ca71a3c6942f21bb5bbe22d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-11 12:54:49.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93903298eb7a3cac1b0ec1bf5cbdc626cb3d1b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93903298eb7a3cac1b0ec1bf5cbdc626cb3d1b8a"}]},"branch":"refs/heads/master"},"0f7fbad9addac22a1682386dbc1fa4daf65d2d94":{"kind":"REWORK","_number":11,"created":"2026-04-19 10:12:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/11"}}},"commit":{"parents":[{"commit":"e7a0a79ee7d39970a1fc025fb4f76b1e6e948a92","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7a0a79ee7d39970a1fc025fb4f76b1e6e948a92"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-19 10:11:12.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f7fbad9addac22a1682386dbc1fa4daf65d2d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f7fbad9addac22a1682386dbc1fa4daf65d2d94"}]},"branch":"refs/heads/master"},"02aeed539590cfa3d3ed1368a677db36cb190224":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-04-25 11:48:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/12"}}},"commit":{"parents":[{"commit":"18c597942c341b54552538f37ceb3e0800195e29","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18c597942c341b54552538f37ceb3e0800195e29"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-04-25 11:48:14.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02aeed539590cfa3d3ed1368a677db36cb190224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02aeed539590cfa3d3ed1368a677db36cb190224"}]},"branch":"refs/heads/master"},"0e4c969b0007a4b144dfc059f1dee63f53e45180":{"kind":"REWORK","_number":13,"created":"2026-05-04 08:17:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/13"}}},"commit":{"parents":[{"commit":"05ae2927a64098a8e423b47a2625433df35db591","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ae2927a64098a8e423b47a2625433df35db591"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-04 08:17:04.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4c969b0007a4b144dfc059f1dee63f53e45180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4c969b0007a4b144dfc059f1dee63f53e45180"}]},"branch":"refs/heads/master"},"8e0693e67d5dc1fb65aa05d4efa35e57a7ab04e5":{"kind":"REWORK","_number":14,"created":"2026-05-05 07:28:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/14"}}},"commit":{"parents":[{"commit":"05ae2927a64098a8e423b47a2625433df35db591","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ae2927a64098a8e423b47a2625433df35db591"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-05 07:28:32.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e0693e67d5dc1fb65aa05d4efa35e57a7ab04e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e0693e67d5dc1fb65aa05d4efa35e57a7ab04e5"}]},"branch":"refs/heads/master"},"5bdd7d2b218d4ead20693cc1dd8af68fa7874a0a":{"kind":"REWORK","_number":15,"created":"2026-05-10 08:04:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/15"}}},"commit":{"parents":[{"commit":"fa9263900015f3bed8a1c78f6c2aa175bd25ce75","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9263900015f3bed8a1c78f6c2aa175bd25ce75"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-10 08:04:05.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n- awscurl: ceilometer unit tests\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bdd7d2b218d4ead20693cc1dd8af68fa7874a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bdd7d2b218d4ead20693cc1dd8af68fa7874a0a"}]},"branch":"refs/heads/master"},"281d816cdabeba45e81b9b40828b2821a0663e0f":{"kind":"REWORK","_number":16,"created":"2026-05-11 16:02:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/16"}}},"commit":{"parents":[{"commit":"86567bb8442151972e10bd3460a32cebc676654d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86567bb8442151972e10bd3460a32cebc676654d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-11 16:02:10.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281d816cdabeba45e81b9b40828b2821a0663e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281d816cdabeba45e81b9b40828b2821a0663e0f"}]},"branch":"refs/heads/master"},"3d8d90cabcfb405782e5c49db9053809f7a4e927":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-05-12 07:22:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/17"}}},"commit":{"parents":[{"commit":"7e7e00dd39cac748838feb6d04269d4e987453e6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e7e00dd39cac748838feb6d04269d4e987453e6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-12 07:22:15.000000000","tz":0},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Neutron issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d8d90cabcfb405782e5c49db9053809f7a4e927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d8d90cabcfb405782e5c49db9053809f7a4e927"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"281d816cdabeba45e81b9b40828b2821a0663e0f","theirs":"7e7e00dd39cac748838feb6d04269d4e987453e6","contains_conflicts":false}},"de548caa21bde3ce1b83627f595f04d078591dcc":{"kind":"REWORK","_number":18,"created":"2026-05-16 11:13:01.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/18"}}},"commit":{"parents":[{"commit":"5d3c724707a7d67900ad4ca03216861b57b59fec","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3c724707a7d67900ad4ca03216861b57b59fec"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-16 11:12:16.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- keystonemiddleware: watcher unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de548caa21bde3ce1b83627f595f04d078591dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de548caa21bde3ce1b83627f595f04d078591dcc"}]},"branch":"refs/heads/master"},"a92fc1c56423210cfac8c5528e013e8cd3ba6c06":{"kind":"REWORK","_number":19,"created":"2026-05-18 17:17:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/19"}}},"commit":{"parents":[{"commit":"5d3c724707a7d67900ad4ca03216861b57b59fec","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3c724707a7d67900ad4ca03216861b57b59fec"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-18 17:17:33.000000000","tz":0},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- keystonemiddleware: watcher unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a92fc1c56423210cfac8c5528e013e8cd3ba6c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a92fc1c56423210cfac8c5528e013e8cd3ba6c06"}]},"branch":"refs/heads/master"},"3b42cbb3413c1a5bb74cfae9efef6186edc1aa24":{"kind":"REWORK","_number":20,"created":"2026-05-20 09:49:37.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/20"}}},"commit":{"parents":[{"commit":"21133bdec4c458ea6817bd6eef816ea1e80b729d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21133bdec4c458ea6817bd6eef816ea1e80b729d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-20 09:49:20.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b42cbb3413c1a5bb74cfae9efef6186edc1aa24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b42cbb3413c1a5bb74cfae9efef6186edc1aa24"}]},"branch":"refs/heads/master"},"36b63a2910c3cfd4a52d0c3148ba26199c3489fd":{"kind":"REWORK","_number":21,"created":"2026-05-21 11:20:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/21"}}},"commit":{"parents":[{"commit":"21133bdec4c458ea6817bd6eef816ea1e80b729d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21133bdec4c458ea6817bd6eef816ea1e80b729d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-21 11:20:02.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nBump python-ldap to 3.4.7 since 3.4.6 was yanked from pypi.\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b63a2910c3cfd4a52d0c3148ba26199c3489fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b63a2910c3cfd4a52d0c3148ba26199c3489fd"}]},"branch":"refs/heads/master"},"2a37197e29d4a93da9c5d9e3437afa7c0c42fd34":{"kind":"REWORK","_number":22,"created":"2026-05-31 17:45:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/22"}}},"commit":{"parents":[{"commit":"9aa84ee59b2c670da31a58ad3e06dc61dfd4c5fe","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9aa84ee59b2c670da31a58ad3e06dc61dfd4c5fe"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-31 17:44:52.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a37197e29d4a93da9c5d9e3437afa7c0c42fd34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a37197e29d4a93da9c5d9e3437afa7c0c42fd34"}]},"branch":"refs/heads/master"},"6ce99873ddd1455e978ec6c1585db61b83fdf3aa":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":23,"created":"2026-06-06 10:45:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/23"}}},"commit":{"parents":[{"commit":"a75e5686ec232952d1656e4ccc2bce371a5e5676","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a75e5686ec232952d1656e4ccc2bce371a5e5676"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-06 10:45:30.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ce99873ddd1455e978ec6c1585db61b83fdf3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ce99873ddd1455e978ec6c1585db61b83fdf3aa"}]},"branch":"refs/heads/master"},"9ff85e452a7deac8a8e3a717537488e774cb1693":{"kind":"REWORK","_number":24,"created":"2026-06-24 09:56:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/24"}}},"commit":{"parents":[{"commit":"96e3d3e18a3f137809bfc3e999c034bef5acf2d5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96e3d3e18a3f137809bfc3e999c034bef5acf2d5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-24 09:56:24.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- pytest: swift unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ff85e452a7deac8a8e3a717537488e774cb1693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ff85e452a7deac8a8e3a717537488e774cb1693"}]},"branch":"refs/heads/master"},"e17a98f45d9693bdc7c106a3f79a946ea303549a":{"kind":"REWORK","_number":25,"created":"2026-06-28 10:11:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/25"}}},"commit":{"parents":[{"commit":"e61b7bc9d919f8d1ee8ea2cc2fa85b4bb751b9df","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e61b7bc9d919f8d1ee8ea2cc2fa85b4bb751b9df"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-28 08:23:33.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- pytest: swift unit tests\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17a98f45d9693bdc7c106a3f79a946ea303549a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17a98f45d9693bdc7c106a3f79a946ea303549a"}]},"branch":"refs/heads/master"},"5dad56dda2df84bfa92ab4db4367ed0816d909a5":{"kind":"REWORK","_number":26,"created":"2026-08-10 09:59:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/26"}}},"commit":{"parents":[{"commit":"64869625d730f243873de5ef5d8e3249307f770c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64869625d730f243873de5ef5d8e3249307f770c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-08-10 09:58:50.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nAdd license for openstack-release-test in g-r.txt.\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5dad56dda2df84bfa92ab4db4367ed0816d909a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5dad56dda2df84bfa92ab4db4367ed0816d909a5"}]},"branch":"refs/heads/master"},"b6801abd08614d96a29e2bea8dc35a4f6e534f4d":{"kind":"REWORK","_number":27,"created":"2026-08-15 09:00:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/27"}}},"commit":{"parents":[{"commit":"57edf8cae25381a37125f22afb3814852ae743a4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57edf8cae25381a37125f22afb3814852ae743a4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-08-15 09:00:04.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nAdd license for openstack-release-test in g-r.txt.\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6801abd08614d96a29e2bea8dc35a4f6e534f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6801abd08614d96a29e2bea8dc35a4f6e534f4d"}]},"branch":"refs/heads/master"},"ec80bfb4b7da9cc46238ba98ec74c04f934f1b24":{"kind":"NO_CODE_CHANGE","_number":28,"created":"2026-08-15 12:48:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/28"}}},"commit":{"parents":[{"commit":"57edf8cae25381a37125f22afb3814852ae743a4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57edf8cae25381a37125f22afb3814852ae743a4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-08-15 12:48:52.000000000","tz":0},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n\nAdd license for openstack-release-test in g-r.txt.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/1000516\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec80bfb4b7da9cc46238ba98ec74c04f934f1b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec80bfb4b7da9cc46238ba98ec74c04f934f1b24"}]},"branch":"refs/heads/master","description":"Edit commit message"},"df1f405fd01e78c9998cd06b1a9995a9e02a96b1":{"kind":"REWORK","_number":29,"created":"2026-10-05 12:08:55.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/29"}}},"commit":{"parents":[{"commit":"7e61181d493536c9a21a9a1d1267a398b0e14b1b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e61181d493536c9a21a9a1d1267a398b0e14b1b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-10-05 12:08:02.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- sqlalchemy: broken compatibility with sqlalchemy-utils\n- pyparsing: due to regression in 3.3.3[1]\n\n[1] https://github.com/pyparsing/pyparsing/issues/672\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nCo-Authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df1f405fd01e78c9998cd06b1a9995a9e02a96b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df1f405fd01e78c9998cd06b1a9995a9e02a96b1"}]},"branch":"refs/heads/master"},"702bd5d0906e786a7d0d80b43ee62d27d406aed9":{"kind":"REWORK","_number":30,"created":"2026-10-10 10:11:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/977551/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/977551/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/977551/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/977551/30"}}},"commit":{"parents":[{"commit":"acbe7007411554ed52482dac365b7dec4d0beb21","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acbe7007411554ed52482dac365b7dec4d0beb21"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 10:28:24.000000000","tz":660},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-10-10 10:09:36.000000000","tz":120},"subject":"Fixes applied to automated upper-constraints.txt update:","message":"Fixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2: Nova/os-vif issues\n- sqlalchemy: broken compatibility with sqlalchemy-utils\n- pyparsing: due to regression in 3.3.3[1]\n- salt: pin to 3008.0 and revert induced packaging downgrade[2]\n\n[1] https://github.com/pyparsing/pyparsing/issues/672\n[2] https://github.com/saltstack/salt/issues/70350\n\nCo-Authored-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nCo-Authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Icde1d76eda9c01cb52c1545c91976efd84b123e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/702bd5d0906e786a7d0d80b43ee62d27d406aed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/702bd5d0906e786a7d0d80b43ee62d27d406aed9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
