)]}'
{"id":"openstack%2Frequirements~977697","triplet_id":"openstack%2Frequirements~master~I16d5745ee5d121b089e374c2eb6d3a840324a948","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I16d5745ee5d121b089e374c2eb6d3a840324a948","subject":"update constraint for oslo.metrics to new release 0.15.1","status":"MERGED","created":"2026-02-23 16:59:59.000000000","updated":"2026-02-23 21:31:06.000000000","submitted":"2026-02-23 21:29:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977697-new-release","meta_rev_id":"9cdf9c3921d7e5fff97a0324a85306cab16bf904","_number":977697,"virtual_id_number":977697,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-23 21:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-23 19:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-23 19:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-23 18:23:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 19:43:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"32589ee3a079861741a50564cf3633d09612c3b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-23 16:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6100facca6934091261a51ef3a4fdced2cb6038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:23:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d384f15b920040debec0dba72752207a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319b35da83b94e2e8089968e71ca68c1 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f571f3561cc4733b66b29dbcc2306c3 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31206739bd8f48c7af11d1eb38f1d939 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96f91790e848479d88c03cf4fda5bfc3 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/53ac65c78cb344e5a4598ec09211b3d3 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14ebd20485464f309cde048e9e71e972 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4780dca9e88d4a9d9694cbefa5b3c607 : SUCCESS in 4m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bbbd975bca7a498abe9c677c7b8ad8fc : SUCCESS in 6m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/44792bc3523442268a82cd953b7fafae : SUCCESS in 3m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cdfe4f67cd60496cbbafa695447fa9dd : SUCCESS in 6m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2fcdb1cdd3f945ee9640b3fcb218051e : SUCCESS in 7m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f13fa0af5a934f8f94cf58b832604e1a : SUCCESS in 6m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/58fb7cab94c648a1bd0215046656b672 : SUCCESS in 6m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ad9855eaed314beaa7d6b58269e35791 : SUCCESS in 3m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e58560cc749c4955ae4b6858c8af25e8 : SUCCESS in 11m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0df6e9229eb14984a78f1486019508cd : SUCCESS in 3m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d3e86b50f344a26b2661f9bfc3abf52 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3979b1e89f7147aca19a999f99ede0f9 : SUCCESS in 4m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/eff77c3c500a4f648b1e4022246d4dcb : SUCCESS in 14m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c0516a4419e40c8beb91eeb1ec00cb6 : SUCCESS in 6m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8b4d2a8974e04689a92e076aa1aeae94 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/263661483bc147edbad61788bd0b30be : SUCCESS in 10m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9ceed4f3989e4dbf8fb15aaa832f1ed0 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0d1aa3be47704e1f9e8cc147ea4d358d : SUCCESS in 5m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3d89016b7cd745368a7a94b0006d37ed : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/801ed83ec4d6441db716155f58bea8e7 : SUCCESS in 6m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/06a9aab25a2740e49b8db618a439a38b : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1d3caa98a426488a80658cb50b4b21f1 : SUCCESS in 25m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e7f76a391abe40a392ff2c04cadfa7be : SUCCESS in 8m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1511573169c94060bdb9a87ef70fb725 : SUCCESS in 14m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2c9f5dd881ab438c8151ae4f6da6a80f : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5110c81adcdc4d4ea59e0c736d77cd5e : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98eb50105e76403797dfd98692a132ab : SUCCESS in 26m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d12cc931e973433fa63f5abcb57b1ea5 : SUCCESS in 7m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1b4e997955244f6bad154804b024201c : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2cead71703484b15aced15c2a17d373e : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/816837a560744cabbd945f593ddab730 : SUCCESS in 12m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b9fe84eba2814be39bb0f48b60cf4859 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e85b7349633456eaad8d138cfeb9727 : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3a269a8248849b79863961477957650 : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9de3bb1a7d3953c4d1a544a0f4981c7ed0147383","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-23 19:43:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a33dab7ab662c76ecb278e2bb7099521bc27fdcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 19:43:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80c66b406ee81cd2e39e992b43deb9faa2e38ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 21:29:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fee199e40dd46ee93dc6ce40b8cb847\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfef202f60a4f248ff4de00829af892 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab804dcedb645069521b553d81afa70 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd2578aea3774169bba9fcdfa0e42251 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f3f55650e3342dfbe0e0112c55e5d12 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eff246adfe454c1f873d7a6ee25f3f2a : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/231a120bbd40455f8a2cfe5347b553bd : SUCCESS in 3m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f502e01d4c2d4fe7ae34298fc1b7c753 : SUCCESS in 3m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4cfb96b2469046cf89a3130b2a2e1380 : SUCCESS in 9m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/49e5b45e32764c29831bc40194993755 : SUCCESS in 3m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/41d830d8a3fc4d2d96120a28679a87d4 : SUCCESS in 5m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/723ab32a6a6a402899eeb3ca47b16548 : SUCCESS in 8m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7bda5b5cf6b34eb9b94535b908f99c21 : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/69f5c522251f46a29927e22edfccc9aa : SUCCESS in 3m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/67f781a96fcc45e68448d0c7b2413978 : SUCCESS in 4m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5f8628e3942e47159689278e57ed9ce2 : SUCCESS in 18m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b824350dbd9a477cb6428d8d0b6e4bad : SUCCESS in 8m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b8543cf6c7364f86bb87bd4edc95f64a : SUCCESS in 7m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a3ff7c44ff8a4c66a4fa16c2821b21e2 : SUCCESS in 8m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/deb188be596c4f8980d42fce782a9ecd : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f09e6cc7ab8144fb9eb2e352c6bc3787 : SUCCESS in 5m 56s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f2b8847e6dde4e75a80ba497ff558b98 : SUCCESS in 28m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ddd6374ed841409cb512aa2050d5c675 : SUCCESS in 9m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/15bb505fac524ad39bd3811b5ab7e1ef : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/72076fe3892740ba9f1e5533edc5af44 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc896986ce154ab4944ab06efa20f2a4 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c8437d48d194d628a5fcca1495f5fec : SUCCESS in 22m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/82d4e5c5adda420ea346fe5434ff8863 : SUCCESS in 3m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f1de8640de1742df942e8d12baa4e90f : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fe306f4ed2f4cb0817ff279200834e1 : SUCCESS in 3m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3a91ff8a04aa4ff6ac06832f413e67cf : SUCCESS in 9m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/04b900bec75a41e3b8b027c3f3575ded : SUCCESS in 3m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5521cc8a3fd461b84f127b694b40445 : SUCCESS in 36m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f10081c1fd142ebb7f42b5105c126d7 : SUCCESS in 1h 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6deaee6d281627c0d8419b0e0a6e08668b680ff5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 21:29:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9cdf9c3921d7e5fff97a0324a85306cab16bf904","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 21:31:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55be05cd085d4b2d826578ca94cc885e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832f863d24ec4c92b9aee56cdeab4a7c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e893edfa4b2b51237541de1e6e75b5df2c5289aa","revisions":{"e893edfa4b2b51237541de1e6e75b5df2c5289aa":{"kind":"REWORK","_number":1,"created":"2026-02-23 16:59:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/977697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/977697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/977697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/977697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/977697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/977697/1"}}},"commit":{"parents":[{"commit":"02ce1830065a5ae8607d7d27725934bb3a8b0fb0","subject":"Merge \"update constraint for ceilometermiddleware to new release 3.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02ce1830065a5ae8607d7d27725934bb3a8b0fb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-23 16:49:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-23 16:49:57.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.15.1","message":"update constraint for oslo.metrics to new release 0.15.1\n\nmeta: version: 0.15.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ia1e8b777654d26a56cf232a04457462590596f64\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I16d5745ee5d121b089e374c2eb6d3a840324a948\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e893edfa4b2b51237541de1e6e75b5df2c5289aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e893edfa4b2b51237541de1e6e75b5df2c5289aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
