)]}'
{"id":"openstack%2Frequirements~977784","triplet_id":"openstack%2Frequirements~master~I1263c3651a3d38c8dd35c0e9b77e196c9ef48f4a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1263c3651a3d38c8dd35c0e9b77e196c9ef48f4a","subject":"update constraint for oslo.cache to new release 4.1.1","status":"MERGED","created":"2026-02-24 10:54:07.000000000","updated":"2026-02-24 15:09:25.000000000","submitted":"2026-02-24 15:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977784-new-release","meta_rev_id":"ad2689252e8e9da6634f6cacb94f3a7a5cb086f1","_number":977784,"virtual_id_number":977784,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 15:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-02-24 13:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-02-24 13:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-02-24 12:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-02-24 13:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 12:44:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 12:53:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-02-24 13:18:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-02-24 13:26:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"fdd2c666376fd113267eb995c5c5dfb0ec7bbe64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-24 10:54:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6889b05a9a4a07401471a0d2448accc88ec531d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 12:44:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0850663226154fc6a6ad44819a027ffd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a164a8fce92c4d3f88288ad291dea646 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1611877cb44783a3eed866d24339ce : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed2497693c9042ea89eaa075b920e389 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f4bb543084d4a03a1bae454507aa81a : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e458fdf0a84c42dc8a87745f130ad5cb : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0d5a34e382545e6a2fdcb3e5a8b914b : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1af8bec0ac634003ab6ee38cff9e632a : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/48845bf5262348d9a34542839d981c83 : SUCCESS in 11m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e90cab6b482b4e9ca8cd818d4cf82ae4 : SUCCESS in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2c36a7d18a8f45b8a4940498f00fd81e : SUCCESS in 5m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0463a319507942fbbb4a10703af25773 : SUCCESS in 8m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a8cddbb82c7a45f090916d2a6438cda5 : SUCCESS in 5m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/69f6fb0c7e654e68a6f920c68dd4bfd5 : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5e4eadb9af204e478eb20b6ff782772c : SUCCESS in 3m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0804511b7b0b4c9d872a5781f8efb820 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7474604708a84928b8b282273b258faa : SUCCESS in 4m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6784da8e4b9140a29d767be366c14fd7 : SUCCESS in 3m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/eb784c87217146bd99737ba393e54dc7 : SUCCESS in 9m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/aec464ea005142da8b800d52cd37eff4 : SUCCESS in 18m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b3d3193ae7c4083b2ec82746d2e5e00 : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bf6140a4ff0044b1bf450149c7a2eeb6 : SUCCESS in 5m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cae564b0324544699bd71bda1b3dd413 : SUCCESS in 8m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c76d4120b5dc41f7b5f65ded7a4b666e : SUCCESS in 2m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f23f4dc355964293a25b635b84990ebd : SUCCESS in 9m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5dc50d1ca87949d5be67a6b81d8e4866 : SUCCESS in 5m 56s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e72488293ccd46439cd715d522311ae2 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2e3434d37db343839c78bd7fda7ea23c : SUCCESS in 9m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/87658ecd53614c91863b0da7d47b47dc : SUCCESS in 18m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc23a6fa9b664470af89541e5d41dd4e : SUCCESS in 8m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0ea5babc0343462aa921a424a9d7eb70 : SUCCESS in 12m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cd8a4e4830fa41ec9b2abed8df5ecd1a : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57a4bdbc16164c549aa0b6d16c8ceef4 : SUCCESS in 2m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adf32a7e5e1c4078a4b7d95b280c0d84 : SUCCESS in 29m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/73385dca27154aa0834a8de9af34f3ee : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/addf4c73b8ef440b9314911cd3e40042 : SUCCESS in 3m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3cb6537472ab477e985dc09899498185 : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d0f0092589384379a26dc01ffe6a2bd0 : SUCCESS in 15m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d25916acb8214eedaf371753258c2c3a : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff110a755d7843cf971d85799ea5d425 : SUCCESS in 50m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c8671ed3f024613a108c8ca32326cfb : SUCCESS in 1h 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"994f27675d5eacb487d6f084b02faa8a07e32dd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-24 12:53:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d6e42b30ad4ee230d5a26a109fecc5b72f6b0f76","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-24 13:18:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0cb1eaad8597aa885bdf331fbbe120e00a47d7ae","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-24 13:26:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c80c1b52369e4894923a43f2646dd0ecf669850b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 13:26:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18cb319e73f68ee6be77e51f899f96de0834ffc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:07:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c13f2d873c4d5c8c6857de1113a545\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35558d3eb9f44a7e9ea080d2b5f6a43f : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dea5b3a30b49529f958066bb8e063c : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b326392fcd5241d99b3db66bcbe11dc7 : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59c864494aab4b60903a83e2e48c5e01 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e5ca02378ba464d8409020c544f4c38 : SUCCESS in 2m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34a302dddb9842ae8504f7632c00b755 : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cfb768f3b7c44a92805199fb6c0447e6 : SUCCESS in 5m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e7a862f52af34859b5b312ff388f777d : SUCCESS in 3m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5f36832416b7479391b8e77632bfaa64 : SUCCESS in 9m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7928fef3797a4027bc929ea537d1e73e : SUCCESS in 7m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f57e6354ff58413c96a45e85d94d1f60 : SUCCESS in 9m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/74b669740c4144a3957085a3cec73628 : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/41f3651e65cd45c994aa810b79a5105e : SUCCESS in 4m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1ffc7d4eb4b844c890522057f9b7ff09 : SUCCESS in 7m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/569246ff5ef144a1b9445f2523f93d38 : SUCCESS in 16m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fceef6ebea9408f9664f1c5f3f74cd2 : SUCCESS in 8m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3dacda06b5de4382840aeb69f8030c8c : SUCCESS in 6m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/277bee38bfa6434a882bdb687c6bfc17 : SUCCESS in 10m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8e006f1a3d18424ca661a2730ea5014b : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2b33fb5acdeb46c3968c89d10f1acd48 : SUCCESS in 9m 53s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b9778cb78970484b96a49e88c368a6cf : SUCCESS in 19m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc7d368691a8478085aed83c2907078d : SUCCESS in 6m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7f9ac6343879475db5af7078f0bd275a : SUCCESS in 12m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5f488cc9557244d8b9efc791b0d7241f : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0cba0fa2bc7433e8a4bd3fa85e9dd9f : SUCCESS in 2m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a39f042a2f3348c5b1dd638fc1a00636 : SUCCESS in 28m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e2182318b50946f3b97092a9e7b02a39 : SUCCESS in 4m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d517c99633c54dbfb5ec45ac88b975c3 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01742efbf25e429093e616fd58c6e6e5 : SUCCESS in 5m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/813863436ba845fc8b3bd8ea2265b09f : SUCCESS in 12m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6b0e86b69e75450db33270499c94576c : SUCCESS in 4m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4aa19eb58fff48b78f1970af9bed476b : SUCCESS in 52m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86c7cd69e7764d04be4006ca6c0e691e : SUCCESS in 1h 33m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d62396a2b7a1ed04d65800bc08098bdd9e339ab5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad2689252e8e9da6634f6cacb94f3a7a5cb086f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:09:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d21eb8049d0d4edab45ae16990b919f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a822a96cddb7472b8f3996859f5ba24f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2424f701e4c741f4c7ddb46e01b2d5b88e1fe52b","revisions":{"2424f701e4c741f4c7ddb46e01b2d5b88e1fe52b":{"kind":"REWORK","_number":1,"created":"2026-02-24 10:54:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/977784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/977784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/977784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/977784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/977784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/977784/1"}}},"commit":{"parents":[{"commit":"af05a42dc4bea4c744cdf26fa4760b7f01fd2df4","subject":"Merge \"Block oslo.service 4.5.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af05a42dc4bea4c744cdf26fa4760b7f01fd2df4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-24 10:44:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-24 10:44:06.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 4.1.1","message":"update constraint for oslo.cache to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I4ddfae9f7bd164831e02c95af438d08f774c588d\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1263c3651a3d38c8dd35c0e9b77e196c9ef48f4a\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2424f701e4c741f4c7ddb46e01b2d5b88e1fe52b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2424f701e4c741f4c7ddb46e01b2d5b88e1fe52b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
