)]}'
{"id":"openstack%2Frequirements~977809","triplet_id":"openstack%2Frequirements~master~Ib0fddc58716b40f6e1cd35213dcbf6a25d5cbce1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib0fddc58716b40f6e1cd35213dcbf6a25d5cbce1","subject":"update constraint for python-observabilityclient to new release 1.3.0","status":"MERGED","created":"2026-02-24 14:40:20.000000000","updated":"2026-02-24 19:14:45.000000000","submitted":"2026-02-24 19:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977809-new-release","meta_rev_id":"d38eae04bd8ba24713281b4a4cfdb27939fc8516","_number":977809,"virtual_id_number":977809,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 19:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-24 17:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-24 17:11:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 16:42:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 17:11:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a81908ec12146f778ed1cffeaec0b6e22f2a5a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-24 14:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352b843c48f93d9dac737fb65cb589343f8bcdc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 16:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a654fbfb05544457bf2c907750f7582e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8600b9cd9e564641a88b2fc4434ce8f1 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee360564c5c4492ab32d5170f605fd6 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/565f059ce0a345ae8cbf30404a69f90e : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b992561988d4da89d61b4d4170607ec : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09eb4e5840ec4d61a1001f32e2a2103b : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e41879cc72a1471fb888ee50e25990e6 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cf8c4fd519c4f9394fc8a8dc4583799 : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e57674269d5a41a687b8de57e4685ece : SUCCESS in 3m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/116b996fedd44929a85359eb8e68314c : SUCCESS in 9m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/54082b5ec4484e53b3a1984ebcccc00c : SUCCESS in 3m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0e618476b43b46cd85b135e7cf106e8d : SUCCESS in 6m 08s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8413dcd22f8346a7a1d618e62fd145b1 : SUCCESS in 4m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0ed372a9cb28433785e1a984a2a0b1b2 : SUCCESS in 3m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/db98c71873d54a4d99e7d2e73284bdd1 : SUCCESS in 3m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2a5954db04304ae1b6c2d70e5eca100c : SUCCESS in 10m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/217b06e97b084e30a7dc0abb6aa8fc4c : SUCCESS in 6m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2a4516adc1d24ab68e20c9f0f581da69 : SUCCESS in 7m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c9398102b8d0419eadef7442255023df : SUCCESS in 5m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f904914fb8534b3a81d8949ee26a290b : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86406920e3c846069d4729f60ce5cdea : SUCCESS in 5m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7624ae94396c4c52afac0301ef220167 : SUCCESS in 10m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c155fe4c223940679113c79bd087de95 : SUCCESS in 11m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/089ecaf0ea294abfb1c9412b40236b46 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/74035d33547143ae94778d3391749a20 : SUCCESS in 10m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/800ef163d4cf4a3c9035526293e95be8 : SUCCESS in 6m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0f22ce9f6bb34fad952b95b48f192eae : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5b788f37820744b482ecaac925489ed7 : SUCCESS in 10m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1d4103fe9c4b41b9bf0f0548b27b1e75 : SUCCESS in 31m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/33e920e6df544d128e8a446eb60dba9e : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c8c3cfba7e23465788210319c7b2958e : SUCCESS in 15m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/437ac7e4922d416d9fa40e95b78f9c15 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75a7377719724a88b9e0050faaefa7d8 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6acb47e556e4343b0b94a7e4f429a1c : SUCCESS in 22m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fb99f50652504d8d855a402c352b119b : SUCCESS in 8m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6697538488654516956668b67b02ce7b : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f63b46d93f3a4faead6477580591fb87 : SUCCESS in 5m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f22a2f32cd02491cb2d0b48979c00ab9 : SUCCESS in 13m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ccecf0e563d441efb0f69e1e96049e32 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c67a05409c0648a99eacc322e8698257 : SUCCESS in 58m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0499f9700a26444e8d289acc6dae25e6 : SUCCESS in 1h 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"212c01c822e87ec99c3a7979f115b2f9ab03b9af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-24 17:11:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a6dae6f35b0473bb838c9a137b1da642df0a1ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 17:12:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17a30cab9c3621bc96b4204cbb0ffa1dd6410650","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:14:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f070e8363a4bbea46d6875f1da1932\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a4a77a9c7346e4b97b28071cb09859 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a329b933714bc388230af22075dd62 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7df6083c4dda41bea7e7777f34db6a6e : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d6f8a0fdbc34e1ba9fcf420bf149e24 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d51a40bb6e048a7906438a39d225926 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57a25ffbc2b64c0a94da8c513c8c90d7 : SUCCESS in 2m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3e7ff820ab5043a4be2c9bfbc62b96d0 : SUCCESS in 5m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f6a45106079e4153a6faa6a5341f80f3 : SUCCESS in 5m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6437c68b03b24b1880e6a5a859807337 : SUCCESS in 6m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6b5672cd5dc44b14b525c656f7e35943 : SUCCESS in 9m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/957f35d28b314d5facf8a256758ef1ed : SUCCESS in 7m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/45f0efcc6e1448b69f8eaf63ac88a6c8 : SUCCESS in 4m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0fe8126b74154118ba0029b98c98707e : SUCCESS in 5m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6632f4ff5ca246e8904fb7de8716520c : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a470dcec549f4eda877da4e81cdffd4c : SUCCESS in 16m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82afd30c70014248a8b6304aaddde8d7 : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/88c8247760ea48d1af5c8353283d5920 : SUCCESS in 5m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b1772d14738545529f88f79f12595bb3 : SUCCESS in 8m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c1cf41a64ff34d9b99f04a7227c070aa : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1e273ed8688641b3b6fb936ead7dca39 : SUCCESS in 6m 58s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0ba4b0ea5afb47a4885190eca7b5fedd : SUCCESS in 21m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d90de74198d943a28d51dd26c8906785 : SUCCESS in 6m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/be0ae993fdc44292b8b1f3e15c043d77 : SUCCESS in 11m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5d64473d562f49c1b990982dcc80c478 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1533bcb8171349f482c8ac95aa8c9e43 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/477ad1209dec45e5be654683c427ef04 : SUCCESS in 30m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/33df806ce28f497cbc8beec6ccedf43c : SUCCESS in 4m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ab3ca9d8c98a4064848fc7e4f1c3c0f5 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ceb580be2f0e448bbb4335cc2848a93c : SUCCESS in 5m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/05fca04467434c8289fac97a5c55ebe4 : SUCCESS in 7m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fa788b31f60542e7a0e3b3c2d2fe742a : SUCCESS in 7m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6a50e6bd09e410abfeb84e37d51dca0 : SUCCESS in 51m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/457566618b0042b49635d0a3b09a0a52 : SUCCESS in 1h 46m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d38eae04bd8ba24713281b4a4cfdb27939fc8516","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:14:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca62c48b7b4b26e650d93730d330b315703bd45e","revisions":{"ca62c48b7b4b26e650d93730d330b315703bd45e":{"kind":"REWORK","_number":1,"created":"2026-02-24 14:40:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/977809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/977809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/977809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/977809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/977809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/977809/1"}}},"commit":{"parents":[{"commit":"af05a42dc4bea4c744cdf26fa4760b7f01fd2df4","subject":"Merge \"Block oslo.service 4.5.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af05a42dc4bea4c744cdf26fa4760b7f01fd2df4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-24 14:30:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-24 14:30:19.000000000","tz":0},"subject":"update constraint for python-observabilityclient to new release 1.3.0","message":"update constraint for python-observabilityclient to new release 1.3.0\n\nmeta: version: 1.3.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I42bfd4f99d1877555923ee23f1a8d91611f3eb44\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nChange-Id: Ib0fddc58716b40f6e1cd35213dcbf6a25d5cbce1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca62c48b7b4b26e650d93730d330b315703bd45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca62c48b7b4b26e650d93730d330b315703bd45e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
