)]}'
{"id":"openstack%2Frequirements~977938","triplet_id":"openstack%2Frequirements~master~Id74ef1ac7c326b9aae96141cc9d71af91ebe5010","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id74ef1ac7c326b9aae96141cc9d71af91ebe5010","subject":"update constraint for virtualbmc to new release 3.3.0","status":"MERGED","created":"2026-02-25 10:38:27.000000000","updated":"2026-02-25 13:03:53.000000000","submitted":"2026-02-25 13:02:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977938-new-release","meta_rev_id":"85ec3756350fbf61787b7646eff35ac328cf6044","_number":977938,"virtual_id_number":977938,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 13:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-25 11:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-25 11:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 11:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 11:53:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"697958d35af82c5b74222a5cff07d25de04b8c8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-25 10:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6abe4d2e716ef9b138745991d6d7ee46b57c6ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:44:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f86a9e6a374278a45dd67653fc7649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dea8acc8e1e4a9a903e95e13c0da9b9 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283a5c3138e44923b74e5d662b2e8091 : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2fa0cc0994c4a8db0ad7349045e928e : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0f6b204384d48dfb5713991997031fa : SUCCESS in 6m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5194f0a24fb044c6b3b5f1bc675ce103 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6f099f74c6d441aab85d3f0063c9a7d : SUCCESS in 2m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7566c358e3424e2e88b6e51c4dc2f204 : SUCCESS in 2m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9d763cd8606411bb30b081ae93b972d : SUCCESS in 5m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66bd2a4094cf4c43a7e0d9bf40332890 : SUCCESS in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f634d9258b6e47768bb5f62a535f0eda : SUCCESS in 5m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a1c26baa45654ba081cee9fa7fb45562 : SUCCESS in 2m 44s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1804b3eab398430a8cc807f6af4290e6 : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bd3608d709ae446ab8d7dbc0acff6cbe : SUCCESS in 6m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/05c4297b84544ba8844ea20b3807aa0e : SUCCESS in 4m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/625e597075a741b29d5f9babc6cdcc51 : SUCCESS in 12m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7f489443deb14dbb9568904ec1cd5e9b : SUCCESS in 6m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c08eeb6daafb4d35ad3e6a2601617a9f : SUCCESS in 4m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bc01458ab723427581687ef189e97f94 : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e9a0c9d7504d465ea2f1530c240a088d : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d47a57358ae4cb08b42b272e05dea66 : SUCCESS in 3m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/16b6040deb8349c2bdb3c5ee17b7d0d8 : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5b4cc1933ab248f095051a10782400cf : SUCCESS in 13m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7b05d40792464a5f859914e80ab8a0f7 : SUCCESS in 3m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ca0783e353684416aa0733df416ae442 : SUCCESS in 10m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9efa34e103d64f0b8abf8448dc05d11a : SUCCESS in 3m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/68bc47fa5d1a4bfc98e85d3c51b8d77d : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/dbd4c7b0bd514baebe17d5f716d160ac : SUCCESS in 12m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3a0874ec3c414008b1ec6e3caf476600 : SUCCESS in 28m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e892c8f397ce4203a9b1d4c04842955d : SUCCESS in 4m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0529d89b057349638110f3246b153547 : SUCCESS in 16m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a03bcbcad5184cba84461479680097de : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9512f80063414b7797d35c2176d3796d : SUCCESS in 1m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53a2f0ccae944005922edf67d53548bd : SUCCESS in 25m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1d2c0799271d41f8bd2317c9e69829c8 : SUCCESS in 4m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/033ae3c4e0d34700b28569a6d38a335c : SUCCESS in 3m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a22333a90b14c179e214fbfa790539b : SUCCESS in 5m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3935653b0fd548358a61b14a09c47a5e : SUCCESS in 19m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f87f839cbba64e87ab7552b78c8bf5f1 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33a3f6ac0e314a8e999b39173b4cb6cf : SUCCESS in 1h 00m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41a5747b6f23409c82a7c1d4eb9f802a : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b1e84b7669639652c771c35154f52e7de0e9c346","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-25 11:53:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a19a7a2d629e2ba052c5e44487ab6682754942bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:53:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"605bcd74e3f642a392592ea112e22caa2661fd1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 13:02:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72c274ca41d4e6fa346cd8efbf5b810\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee8fc46f7174792bc99f279ad7f5a45 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30df471f279044769cc81a9699f777af : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0334a42a64bf488a85ea7be43480ca77 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/331873c18b49476bb1a491d01f4b3568 : SUCCESS in 2m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa1357f45fbd4a89b47dd0ae0256bdaf : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c64919c03bfb4f52b83580473565e597 : SUCCESS in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/352caa4580054ed5b400d21b0777c8d8 : SUCCESS in 2m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1d47313b039640079367733d4b2d3834 : SUCCESS in 4m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db9b042d85dc44a58ae6eaf449f12966 : SUCCESS in 6m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a04e8935b00c448ab0cd99fcc3c7d621 : SUCCESS in 8m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7ca37ec86a904cbe90741db4ab105b2c : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5ebfff15a35043bdb88d2e91eabd3d29 : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/611108412a554ee7b41ed1ac4a660581 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/719fb55b01ef4cd79b71e51094beb769 : SUCCESS in 4m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3ed7189cc1174e53be87775193aec184 : SUCCESS in 17m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aec0e4fab89a401eb28062a05aec815e : SUCCESS in 5m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/12617f15b661437783a65b3c7c08a398 : SUCCESS in 9m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/baa45180e75249db8ecbb76ae9615512 : SUCCESS in 8m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c9760a1b8f2f4b5f881940856177ed5d : SUCCESS in 2m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e66ef79fc9d945fbbea76912df703835 : SUCCESS in 9m 54s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b65e6a33bdb5421abe5edb50dbac2527 : SUCCESS in 23m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6da743bf71f1402786bc1c5ffc3084a9 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1ee284edc2484f0d8a083230c3c032c4 : SUCCESS in 15m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/592cff42026643bba5c983c12bab0e67 : SUCCESS in 2m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f87b25ebb064283827c3aa3daa17c5f : SUCCESS in 2m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/62a598c4f5e146f7b139acbbe582e639 : SUCCESS in 29m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3cb946057a624439910fb69365883e62 : SUCCESS in 5m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3756e5e9e64a42cd8f920e09a0c62ea1 : SUCCESS in 3m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/802a13921a5c4c1abe4ec020dca3fc2a : SUCCESS in 4m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/83b3189fddb44a16b5881b42e98e944d : SUCCESS in 20m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/58bcf8f0d6e64f728c03d0dca33ae462 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf481ee542074a70bef72f2b98941df6 : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e99e0ab281c41459cf66c7e6d963dec : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"be0bf70d86d117052f98c93aa325c0ebaa01aba2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 13:02:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85ec3756350fbf61787b7646eff35ac328cf6044","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 13:03:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08b2915ea89f46239061427d27d8f285\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97170edb4b924b8ca6a248404f5e2559 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0103da572def4f9e00415a46b1222d2273dc3564","revisions":{"0103da572def4f9e00415a46b1222d2273dc3564":{"kind":"REWORK","_number":1,"created":"2026-02-25 10:38:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/977938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/977938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/977938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/977938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/977938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/977938/1"}}},"commit":{"parents":[{"commit":"c293d9ae28b37958eaf9bb1d7dc5d03c6f602411","subject":"Merge \"Bump infoblox-client to latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c293d9ae28b37958eaf9bb1d7dc5d03c6f602411"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-25 10:28:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-25 10:28:22.000000000","tz":0},"subject":"update constraint for virtualbmc to new release 3.3.0","message":"update constraint for virtualbmc to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I17fa718919109f2dad012277b7165d63e1bb7bbc\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id74ef1ac7c326b9aae96141cc9d71af91ebe5010\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0103da572def4f9e00415a46b1222d2273dc3564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0103da572def4f9e00415a46b1222d2273dc3564"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
