)]}'
{"id":"openstack%2Frequirements~978193","triplet_id":"openstack%2Frequirements~master~I300b1b6b522a9d79dcf53573daea67f518f4985f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I300b1b6b522a9d79dcf53573daea67f518f4985f","subject":"update constraint for python-manilaclient to new release 6.0.0","status":"MERGED","created":"2026-02-27 13:28:40.000000000","updated":"2026-02-28 00:22:10.000000000","submitted":"2026-02-28 00:21:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978193-new-release","meta_rev_id":"9ceef17b6954ac380ef85963da2c733984669a75","_number":978193,"virtual_id_number":978193,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 00:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-27 21:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-27 21:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 15:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 21:02:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"76d4993e3373b381e3e1a95d16f4d9dc786d3706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 13:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"104dac3ec951cc4aed36fea85737d7acb4081d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 15:24:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3eb879a75b8e413cb7a5de5cb4c18463\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4630812afa39461d844ca6b6d932cd16 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1298b556301c46b69d8a4c8f80241606 : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d054060645f4b4b82c294f8283ceea2 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52fc955c087946808501cd58212f1bcd : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e5766359e164ec089250ccbda15c670 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d7b072c48124a94b7b68248bbda844d : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aeb42d850de84a7da516f980ffaaf740 : SUCCESS in 1m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f197c916f17546a2861ecaf1d93e5422 : FAILURE in 2m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3ca1ac1ad4a44cfe9708944997fae0fb : FAILURE in 3m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f29d06952d0e481287534a4c73d98e8e : FAILURE in 2m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7f5b093fbd4b48a596010246d45762d8 : FAILURE in 3m 15s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0b063c9565344c9990a1c25e4b82ae1c : SUCCESS in 3m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/112e5174ce80438eadc7620126849ff3 : SUCCESS in 4m 41s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9dd034be743e49c8ab6d54c0ee0e8ecd : SUCCESS in 3m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/612fa017b5264709b672940cbf5e3116 : SUCCESS in 11m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b6c84c8e6da84ddd856b896bfb93522c : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b894be5bf59a429196c140780a4261e9 : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b4cf5766ff0c42c8bfed3d440aa4369b : SUCCESS in 4m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0309b781643a41b7a9dec4dc6e726943 : SUCCESS in 16m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bb5b336d6734fcda78a517ac297cdbf : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c8478d675bde4443b482bcfba1b635bd : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/441ac5de14e14c3b87b2f58a4d8535e4 : SUCCESS in 10m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/31f85a5c7bba44ffa5006ee969ed2c8a : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4be18cc3aa824040a5e5227651826c26 : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3bd0a127ab8d4c828daad81452ed882a : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4f97199ebf0b4e0ca30b65e2b363af35 : SUCCESS in 2m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8c35445d722e47b0b0f04de2ad988014 : SUCCESS in 11m 33s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e78a5506a9d4598858226573747d966 : SUCCESS in 27m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f1258e09a8464517b7f86b9fd43759ec : SUCCESS in 7m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/237e24b69158453eb0a6586069c1578c : SUCCESS in 11m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b8c71f24bf9e44aba288d1e6b7c17c3b : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f58f2382f611406db2cbe1e9ae9ad92d : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/edbf7b8ffa1c4a0ca3559a6f3b3685c9 : SUCCESS in 26m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2733d785375b4ae79f939ca8eb7f706e : SUCCESS in 4m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d75d1cc81d4a439688867c16dfd27cb2 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8536adc4e2f647889f5402cc70dc26d3 : SUCCESS in 7m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f8452814cda54887b4997695ee0fa27e : SUCCESS in 6m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/752887a50d8241ab81e8e7f25a9750b5 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1de4580f5d4b49239accd9050cc0f181 : SUCCESS in 1h 01m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9b5192c77fc4990bd77ad7b6dc24a8c : SUCCESS in 1h 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b31bb2e1cb38b4f97d2db9fa75d54ae37f64566e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-27 21:02:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fe621a8db1c28fa9fe1b21015b3096f827c68e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:26:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c0a92a3f784ea290ca2b8ea09143f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbd1391425049c7a78b2e6a9fca5841 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7fc36ea4a54f8392797385b99fc86f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f23c35c2d524f8091a96ed7c39cc175 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85e54ea328494f63a17c3190f4b8b2ab : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0986cd4524b245f0a2f6f1cde604d24b : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c81efc27334b4351a39a78e0f8829b09 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ace5d1b2bc38473dbee638c4a4cc60a6 : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ca3461dfab948e4b4c6a6f6ad648970 : SUCCESS in 4m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32a4b21d3efd4f11b9181cd04fecddb9 : SUCCESS in 6m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc5bdc53cbc14e8da233853e09a7a7d9 : SUCCESS in 5m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e0d32286da54d3caf83cd6e97fb6c46 : SUCCESS in 5m 20s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/50b897eb9eec41f28180a78c584762a3 : SUCCESS in 4m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ec7bc6ab134948b3b5c01ed8df049440 : SUCCESS in 6m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5863264059e945789ca28a50535658a8 : SUCCESS in 3m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ce02855e3ddd4be6a839b30cedec22f5 : SUCCESS in 11m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/29f190009c944e5e8cc3def015c39dba : SUCCESS in 6m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/617c4f0ac25546ca90270b796bee9561 : SUCCESS in 6m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/589f1eb469b84e1089d98d380c32ebd7 : SUCCESS in 7m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/35011e8d2c434242a01cdc8526cea974 : SUCCESS in 7m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3656e7e1c388419da1cd58810963a7e1 : SUCCESS in 4m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ef9583cf29b14f58b2e89a988d55e3f4 : SUCCESS in 8m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d002e2480a34402598fac33e76e8ba92 : SUCCESS in 12m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6d35968d7ab640ff8367e24d1d4ad60b : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/939dda4ab84843bbab0a87b1d8fbd659 : SUCCESS in 7m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e537eeebd11d41d98a81d2e99aac0b2d : SUCCESS in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cbba73cde3104e2abea8fe323d1f735b : SUCCESS in 3m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3649ac5a1b0a456a8a9c315e91bea65a : SUCCESS in 9m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fe438e9aefa14b40b4afbddcc708def4 : SUCCESS in 26m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/74e9d24903fc4474ba3d05927bed51d8 : SUCCESS in 4m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0acd8764067543f280e1dc1a12e1042f : SUCCESS in 11m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8a15f7a5b4e84ae59be7db4de93a614f : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b938c690567a4f7787539334c7bf532f : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c39a5d5574644415bda74dcda8c79068 : SUCCESS in 26m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ee97eaada3db47d5bceb69a615c5e250 : SUCCESS in 4m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/545fdcef16e64416b14703c0f21bda3a : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5f23c1b4a97401b9227150f3a43c6d7 : SUCCESS in 3m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3ce56d49f63c46838180acb9a5a44872 : SUCCESS in 17m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/07f0e6fd8a924d34938ec61973be6282 : SUCCESS in 3m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f0da5a785894b0bbcab9771a75c3329 : SUCCESS in 1h 02m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f1b752cd8e348e898482c08d8064d58 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"797fa61b77622f610178fafb3ba5f8ffaa9a5ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22493a6f2c0b650551f4857b8b008d9a20187f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:21:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c92635e556d4a78b631de9f1f61deb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0cb02631344cb1a0df10b7972f9acd : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a468ead2cae48d0955b1bec9909abee : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cc30584628e41429464b97dfbee2aac : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/798a146f7b6d4e669524ff4c66c2985d : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c629e4e398146749dc98027640d293b : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/813886087f6b42ebbb3c2ff1c62855fb : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1449312a8eba40c19ba3cb8a8987935c : SUCCESS in 4m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98026c68ae9a47a6ac56f391588ee36f : SUCCESS in 8m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ed8daab63d1046efb15c4e1703a8cc68 : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/aa5a5dacbe8c4209af294348a532c8ee : SUCCESS in 11m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b133dd35d7284ca788fedca30b94cd99 : SUCCESS in 9m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/65643ea7f3b3402599d90096e0765ef4 : SUCCESS in 6m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a9b13b538fba447c8baec0c50ea7efac : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4fa9c2b137a24b089792a44bf7fe0b9c : SUCCESS in 6m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/15e55aec59b34237a5ba332c3e7f2c1d : SUCCESS in 17m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0439018463343dab6c1a10d0d4ea7d1 : SUCCESS in 5m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a81dfaf19585473e97d3ef0184feef02 : SUCCESS in 8m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e7956790bab44483ab63ea9bdfe22b1c : SUCCESS in 12m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2ddb7ad0c778412c8a724b5380415c5e : SUCCESS in 6m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4ecddfda6bb047e8989074d0724d849e : SUCCESS in 8m 53s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/df4fab1f738f425fb887ae4cacdb3f8f : SUCCESS in 29m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b8109b1f69054da8ac5f009d9e27d0eb : SUCCESS in 9m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/233996bdc29f4c598b77aeb723aa68c4 : SUCCESS in 15m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/64660c5e5ff74d7a9e82ebecbda5ac0d : SUCCESS in 3m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa4279d8cbb349d7b13ffd6083c25e6c : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad52a088665549589dd881f9cc307861 : SUCCESS in 29m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/eb75c85834934e7998dcd48331cac5ae : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1f9c62497c1a48f7b825a8dfe4bf1475 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ba0905c00734b99a62bd0e41a3ea50d : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/28a82aae34fb4d27bcbfa8cd76ab4533 : SUCCESS in 18m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a2bbeade559141418e943e089e4bc985 : SUCCESS in 4m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0039c9790f4f409a9f78e3ccbfb2ade8 : SUCCESS in 1h 04m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/138c6c9bae5c451fa1dca287e0609cd6 : SUCCESS in 1h 49m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1185de745527d27aacce5d671824dd1a9714c5f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:21:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9ceef17b6954ac380ef85963da2c733984669a75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997e9df428ba4ffb9b4102dc354e028c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e059f231834a484e9a639e9fc929412c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adbfd6d8e4ffc550a322d49258fb10fa2a297964","revisions":{"adbfd6d8e4ffc550a322d49258fb10fa2a297964":{"kind":"REWORK","_number":1,"created":"2026-02-27 13:28:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/978193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/978193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/978193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/978193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/978193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/978193/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 13:18:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 13:18:40.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 6.0.0","message":"update constraint for python-manilaclient to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: team: manila\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I781ac1a9a37e40b3c2417ddedce39df0f5baf93a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I300b1b6b522a9d79dcf53573daea67f518f4985f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adbfd6d8e4ffc550a322d49258fb10fa2a297964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adbfd6d8e4ffc550a322d49258fb10fa2a297964"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
