)]}'
{"id":"openstack%2Frequirements~978200","triplet_id":"openstack%2Frequirements~master~Iad55544b43c60d98c8c8cf5fc0337cac4c027bac","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iad55544b43c60d98c8c8cf5fc0337cac4c027bac","subject":"update constraint for aodhclient to new release 3.10.1","status":"MERGED","created":"2026-02-27 13:49:21.000000000","updated":"2026-02-28 20:22:34.000000000","submitted":"2026-02-28 20:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978200-new-release","meta_rev_id":"850b6ca9d93d3ef972d08c5789af4756b5a95a73","_number":978200,"virtual_id_number":978200,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 20:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-28 13:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-28 13:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 15:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-28 13:21:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2b34fd4ff43b4b3fe2fc3cdc88d781e203e25169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 13:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f901855ae0de3f5f628a73f695f5c38e9f825359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 15:43:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11868cd8310241358d81253509d628c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810106d70ae2484d88bf6c6538f556d4 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ca31ea0b334a0d8ec9d1625ebd64d6 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4b6865749d845079e806a273e6f6803 : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b4e6128b37048588ad1fcb93219cb8f : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1de12c1873b844b18f95c4927abb6cd3 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74f4268664bd41d78784b9798a6e722e : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/008b2dd0a853490ea83308e29c6dbf3d : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cc98d813d04f4f67b0ea4c6d1ac3d051 : FAILURE in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a1f8a73cf7dd4ee5af3653fe315eba2f : FAILURE in 8m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/41c3c178d7ad4f3894e051b8ddc782b0 : FAILURE in 4m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/891dcaee19a34f3cb898d9d5bcc101ed : FAILURE in 6m 38s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5f5ebaec5eca4f57a458a7b5f5a10b11 : SUCCESS in 6m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ebc928725f5c4a42b00af4e6aa3fb633 : SUCCESS in 5m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/25e938f1a74f4196ac19a70d6aaa6b89 : SUCCESS in 2m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/98789b8b75e24db0a4b4e31e4c39cd7a : SUCCESS in 9m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fe18207bcd6d42428aeb46b77b2eda96 : SUCCESS in 6m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fbd89c981cac44ddadaf041c2d43ce57 : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a03c557dc60c4b7583df9bc5a1a2cd19 : SUCCESS in 3m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/533bac012b5949b3b857815fb5738e03 : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac89b488d2954a0b87cde068ffeb2c29 : SUCCESS in 6m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dc1f41aab8ba457583f39f3eeaa94ecc : SUCCESS in 9m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/53ab5207a8284532a4c12b9c31315520 : SUCCESS in 14m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ae2175e50a28400c9057ca70e68ffa51 : SUCCESS in 3m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6802cf2de7324619bd40503b4d7a0ab5 : SUCCESS in 13m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/950fb60918c14cbaab576aeb4a0ee52a : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5ecba6a4620a442dabc151935a40bc15 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d21466663ff8481a8e5f1858ef99f86a : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c58894c6bba14183ba64353b84974169 : SUCCESS in 23m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ae4b197c2c364510aac1556bdb951338 : SUCCESS in 8m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9cf07715afe24eed94b6d6379dec5731 : SUCCESS in 12m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1aa4f407ecd04c239efbcefd64902e56 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0bd92d234c65444eaac90b4fba8a1afb : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be0b9275a9cf41b29f54c20246fcf528 : SUCCESS in 27m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/21c50122332449a79eeada729fa4849b : SUCCESS in 6m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4ef5c5dc8b644c588eff8c9ac234aa38 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d65bf2eeee4340248b48707073fb0d8e : SUCCESS in 6m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a18e1ceb571b401382eac02173bebc70 : SUCCESS in 18m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c119017d9af04301979c04aa0af4be4e : SUCCESS in 7m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04184a6a390446a49d103495f31e86a1 : SUCCESS in 1h 09m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63a3aadb38e541fa9c63fe3dc4e52aef : SUCCESS in 1h 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3381d13fabb16043cf2acc41f059fd3d6cf5abfe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-28 13:21:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8b7f415458cd3479078a853019cbf2079c0237d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:33:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a01e60585cd4ec4923367f8630a4b27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5420742fc844eb6980aa8e477362e2f : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650e8f299797433696e2d7bbf1c1daed : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fe7ed5f5fd54cafbcaa9705207a1172 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2b752bfc7164c909e060b56c48f6e5f : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1f1b5a966ed4cee9f875b73949a5edd : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdb3c6da49df4b908fd27ae5a34d602e : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f7bdce86f8849609b788d156aeb129d : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1804d0e0a8eb41b2941b44447f835076 : SUCCESS in 4m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e91842cd137d4843aa0cf4738d4e9854 : SUCCESS in 5m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bc232288cb814ebab1dc19ebff4fcfdb : SUCCESS in 5m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/92f784c5347547cca258545530fe8764 : SUCCESS in 5m 22s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4c89894a895646649602efde6942436f : SUCCESS in 5m 22s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f8bf954f232f4fd8a61f8626b7af0f79 : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/85ad8f4c3e3545f6a77cc94d81c66a69 : SUCCESS in 2m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/88e62309860149868bc319538375b4fb : SUCCESS in 9m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6ddff6d376424d5aa0420618ac910faa : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/812a1bd1f38f48d98fa9d446b5216c69 : SUCCESS in 6m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a9431cc048974b269e8f948724eccc36 : SUCCESS in 7m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d2d67ac76172414b930799e89f996704 : SUCCESS in 16m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e4e242ca3424090a63bd5de44338d1b : SUCCESS in 8m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/344edfd7980e461abb9548cab73190aa : SUCCESS in 8m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/99146cdb63374f21a87b4a070976fcc6 : SUCCESS in 9m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0f1f5813b67449a2b106a567d0be9ed1 : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b6241631166d4274aa1ca601c39824ae : SUCCESS in 10m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/774c9aae2ca243d79fe059bb04dc46ba : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4e565a8503f94d1da8b35142a77424db : SUCCESS in 4m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3de273173c574d3aa286a1422feafd64 : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1c461f9f5bed4d9dab61b0e5a4707530 : SUCCESS in 26m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a2a25377b6d24b1ba5f4a979e004d264 : SUCCESS in 8m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e6e3ecf5fd8a459fb03ecada85640426 : SUCCESS in 15m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2b00a318445944dca5eb7853c14e091a : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2ef8eda25ea43f69401b2d4ca2fe2eb : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d418d5016fe4f3e897a4fef88b3295c : SUCCESS in 29m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2999c74e79d34f00bf944a0e55b0fe9f : SUCCESS in 6m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f72e98993138443a80f2541d0c291696 : SUCCESS in 3m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c256b15bf98a49c0acab4a30d1f582fa : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7bd2c4e798ce4ee991188b9debf693db : SUCCESS in 12m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/831c335558e94d75967a4aeda7968bba : SUCCESS in 6m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a596de4275314615a70075c0d48005b0 : SUCCESS in 1h 04m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd1c133fa0534ce5bb5ba3a170ca50bc : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c2c56051f388bc75e80b104f80a1e57718ae9130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:34:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"811a18fe7e3e53a83f120f9d5f7129bc3002a7e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 20:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e124683165ea4e069600030fa6ab70a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ef57ccbfdf482c951ec84387165c1b : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f71cc5da70f490c834c9bdc1e6e802c : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c2cf64025054421b759d3b82cb8fc64 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80e132c24ea0487ea2487ec06ec94c2e : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/857e1dab50fd43f8a57dba7041913397 : SUCCESS in 2m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a713116a00440d2b6c3a720e0d82d7c : SUCCESS in 2m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2a3e2730fe2e4232ad70776e80bc3309 : SUCCESS in 2m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c57dd954ae74a0aa2416c6a37e26dd5 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/62d49a95188a4de39989c336cf5fb8f1 : SUCCESS in 6m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e5cc02be596c428ca0ac2fc947090415 : SUCCESS in 10m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/93592e4dc4084b16b445b4f9afa3951f : SUCCESS in 7m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1855679d908c41ada8a89b428cfe980e : SUCCESS in 5m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0c8e7cacdcd947e49fb271fe81c2d3b6 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2426017ccd124009922a314d102b6258 : SUCCESS in 4m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2b884cf1d22845958f7b811a46b20efe : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4097e2fa28ea455a9f99d893211c4f37 : SUCCESS in 6m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/580e02e4ed9c48b2b855007f464fb1dc : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/476e300f76574819a302e8770c806359 : SUCCESS in 13m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b48a2f0333be4568b8b247f2b6c0d6a1 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/94e8217f7d01404cb93f0773e839b20c : SUCCESS in 8m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0ebe9864541b43ecb9a0643ac43a81c4 : SUCCESS in 18m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e16d55be14434e7da3279374a1d53adc : SUCCESS in 4m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/366931e0648741f089d5b3ddd3d7f17f : SUCCESS in 16m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4ccb16e6a16243e3b66bc1aa34c80f30 : SUCCESS in 3m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/918403e19194401dab11b036fd504818 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af0dc73075894d048f071c3056b5810a : SUCCESS in 29m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/541af2831e8f49b9a61dec2feaf2b193 : SUCCESS in 7m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0e771ec40b0b4b4381bf0a9f804d319c : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a198683ee62d46daaec820912d89abd8 : SUCCESS in 3m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e6770082d201446bb98a8d0dbc491d7a : SUCCESS in 22m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dd6830184a7448209f148f572eba2c24 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a48c0134578471fa7deda45085f97a9 : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbe31015edc94bac8d585a0cfa83c392 : SUCCESS in 1h 42m 03s","accounts_in_message":[],"_revision_number":1},{"id":"abd3f9da41b8da00d56db9a5e8bf5e29029d2113","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 20:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"850b6ca9d93d3ef972d08c5789af4756b5a95a73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 20:22:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fff6b14e4d84f49844a9898d564f5f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b46b82928749c9982408cb132a357f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdddee2f80d8e16398564ee6aada4cc98ae0cd4e","revisions":{"fdddee2f80d8e16398564ee6aada4cc98ae0cd4e":{"kind":"REWORK","_number":1,"created":"2026-02-27 13:49:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/978200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/978200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/978200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/978200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/978200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/978200/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 13:39:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 13:39:20.000000000","tz":0},"subject":"update constraint for aodhclient to new release 3.10.1","message":"update constraint for aodhclient to new release 3.10.1\n\nmeta: version: 3.10.1\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I240bd634d03ee46d810a7b3b6035c2fa66a96f44\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iad55544b43c60d98c8c8cf5fc0337cac4c027bac\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdddee2f80d8e16398564ee6aada4cc98ae0cd4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdddee2f80d8e16398564ee6aada4cc98ae0cd4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
