)]}'
{"id":"openstack%2Frequirements~978217","triplet_id":"openstack%2Frequirements~master~Ic102aa55f3460b4e03455d36c2d877243c38b9d7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic102aa55f3460b4e03455d36c2d877243c38b9d7","subject":"update constraint for python-vitrageclient to new release 5.4.0","status":"MERGED","created":"2026-02-27 15:32:48.000000000","updated":"2026-03-01 09:40:34.000000000","submitted":"2026-03-01 09:39:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978217-new-release","meta_rev_id":"24b23128f38b02e2cd7c84fbdc97258923014a91","_number":978217,"virtual_id_number":978217,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-01 09:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-03-01 05:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-03-01 05:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 16:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-01 05:49:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"14273fa96ccae77b462ba836f1d88540e9d300eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 15:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"938773e98c2db685c31d349e78c41f731f344c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 16:28:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/879e1865ed0b4c2a9464d69a04abe04b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9efb294c334d5f89c78dd5fde7e1fc : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59212763f71f43a6bf357cd180b037f6 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea3809c446014b3080be7f0482101dd7 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9307ce2364d4e86bd36a4d9897187cb : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ef050d98db48489bbb43df0f47bbcb42 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0d59bc76a18412ca47880cd95c9812b : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70ee3b60aea743469f09104e8d1257aa : SUCCESS in 3m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ced30da95faa49a696cc2ebb1fe9132d : FAILURE in 4m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0815ce0ed543463c9eb91cf057b559d2 : FAILURE in 2m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/af99ec830bc0449cba1ca3c5f21d762b : FAILURE in 4m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9830dd653cc24f139fe82975a12fcbb9 : FAILURE in 3m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9605c26156c04944b0215305b77368fd : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4e52a0d1ddd749a6bf71b3bd177b3bcd : SUCCESS in 3m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7c09f71edc7a4f8cbd209edeb11021a1 : SUCCESS in 4m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/298dc57d3bae41ef8c02c5d564aa5a53 : SUCCESS in 11m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dd838a0e51b141f2b0574bf2f0f664d5 : SUCCESS in 5m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/264d6f48aee44259b6086aad5dd9fb8a : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b90ae43586944d7a5d46fc66c4a4688 : SUCCESS in 6m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bf7fb4b6e1e84ced9e6e377b1f06ab58 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5647925fc6a44b7f978c9a36813a073b : SUCCESS in 4m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/308991fef3894ae18283ea0e0c9344df : SUCCESS in 5m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b4d20eeda6b04a8886d955b89d1af7fd : SUCCESS in 11m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/771d1577abea45ff86771fd8ec744862 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/facb668d1edd45a2ad300d1495cae316 : SUCCESS in 9m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b524a51c913c42479567de6c22ecd4c5 : SUCCESS in 5m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1bd2d8d82cdf4188948dba39a66ce88e : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/78d16008622f4166936bbe45ead5c006 : SUCCESS in 11m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/024918fc54a54ea4b342aee750cc5007 : SUCCESS in 21m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1171cd8afc8941848f8744316f02dfc3 : SUCCESS in 4m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2b9af45f514b4b478a0f97f65f0df9d4 : SUCCESS in 15m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0293dc8a05df45c8a36778f5914cfe05 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a0fcbe6c6fe4ad3ba99f830116f4748 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c99c7f68e1a48938cba758dd081642b : SUCCESS in 32m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ac79086f64e348488232a00392e0b021 : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/139c4549754e40ad96a1a20fd7f01b14 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2010f0652d0641dcb505cd1cf8d92ca5 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fe46a8492d7141e2a44bab889065056f : SUCCESS in 19m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/38b00f13572c4c9f9793ed698d3f7874 : SUCCESS in 6m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/822e6b0e235b496bb38c1d61d6288e15 : SUCCESS in 35m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/660e7b784eef4375a8396a5de5190cc2 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"878d4b1c2e9e4bbe1bae6ed6129f0852b0c2d4ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-01 05:48:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e88da6b532d57d1ae9208d9c2efb29cd34a09b98","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-01 05:49:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5cc53c21a77ff530fae1111b0637798a5e46690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:48:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e9bf48b3084782950530287266ca72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31d0e5a37e24d25bfc3787688bd9733 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74040e8b554479d95db8e0c23ed7c9e : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/960ec91cf916452b9f266985f20cc81a : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a684942614bd4632bb771637e3639bda : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b24892572ff4b9e9afbf9271be5ca58 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b84f89b755884dafafbe7776a78b1a08 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d0b5fb0fda44adc9375f78048610e83 : SUCCESS in 3m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/34582f850f0a4dcaa8ac5c1db21690cf : SUCCESS in 3m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8e674d7ce6704a08baf346b32d658fa3 : SUCCESS in 3m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d82477c9110246cba385e9aeb615f4eb : SUCCESS in 9m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e1e4ef4b6e764ca7a8bff052571d046c : SUCCESS in 5m 16s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/35c7ea86bad440cd81f68a00b62c2b21 : SUCCESS in 4m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/49addeb855ec44c48e4fa51e5c002ef1 : SUCCESS in 5m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5446c8c98b0f4b36b20d9f9e71a4b6b8 : SUCCESS in 2m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6d754fb8ef2a4f6ba6472925a52a2cfd : SUCCESS in 7m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/27321b2eec274c5ba2c0c9c8cde1aed6 : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4e8f3d70dfb443a79f46154e6b7da408 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f54740bf3a264e86ba7f0363aa423d60 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ca7d810354dd4a3791fb97626f780a34 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2494277ad96a4408aeb6eb33182474d0 : SUCCESS in 3m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/da5d7052d05d4e6c989e0ebfb6d2e1d9 : SUCCESS in 5m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/81484a1e120d44439816af0539b2d6c4 : SUCCESS in 12m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5403c7c70af2458683c0a0f98f462842 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1ec00c7f156e4cf1970a91c6a63ef534 : SUCCESS in 10m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ebc3545a085841ac89d71bca8406517e : SUCCESS in 2m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a61d9f314ed843bf875437c62a52ab91 : SUCCESS in 4m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8de69dffe4ae460f90b0a147e7ec88ba : SUCCESS in 10m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b87559067d234aa9a6f35be3ea401ea7 : SUCCESS in 28m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4aae122a5a5d4c3f9142087bf6b9aa4c : SUCCESS in 8m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/43f70f81debe4414816b31ce9d0cb9d8 : SUCCESS in 16m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0ea94323177e4bf69e8b4cf7207f32f5 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ab6c7d3b9bc4b179554b07ea54c79a6 : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f6659436ea84b34ba3cabeb42855b3f : SUCCESS in 28m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a14e61ffe7b049298c1d6f61a57f54c3 : SUCCESS in 4m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/06ff696cde0c4cf7906fd54e3da04984 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47b56afa124f4de4ae125b378dc8d215 : SUCCESS in 7m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1485eb67911d471b9ee270b354783a03 : SUCCESS in 11m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fc109be1e9524a8d8ec62b3d79095aad : SUCCESS in 5m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b754a3a0f58942468f6127b5fc73e311 : SUCCESS in 1h 01m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6013e73742f1441aae8c00dc215ad425 : SUCCESS in 1h 55m 11s","accounts_in_message":[],"_revision_number":2},{"id":"7c81cb1fd01fdfe8ee02046570f4011b67f6891b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:49:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d1f90ddf5a3cc7a76d0481f4c365cf2d5e00feb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 09:39:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2f04148fd64760957a0d11a39a847f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49db33a1e784bcab76ed166192d108a : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c68ae6d25824fef9c28ac42cb4a6d56 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e46a0ae922241e09f2f8d82b995a135 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7420d2f58eb04940a1f8b251b116a9da : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3782b97156440b99f0be95309173c29 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1be196107fce41ec9e0d2d6c16a32cba : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5eec9e0869f3498ba95e2536168f8867 : SUCCESS in 2m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/90112799e15c44efa1681cec5a0d0b55 : SUCCESS in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/451491c113d641168087ba5c91db0ee9 : SUCCESS in 3m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e222135467774e3a85646d74493fa129 : SUCCESS in 4m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1552ededec6b4409893dbeb5e38461e8 : SUCCESS in 7m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1b8b73382a5848068329b6fd721881aa : SUCCESS in 2m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/300bb34d3a1845a9a211db5868021658 : SUCCESS in 4m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3686698708b8441b90acea3f2ce2c27f : SUCCESS in 6m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/34c14838eae14f52863edc66bfcd24ed : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59993f36ab29442ea06342462a4eb8e7 : SUCCESS in 5m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e1ecf655734b4922ab8e86e6e2a5b15c : SUCCESS in 8m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/476fa9850f164a03a5651c1b4f42141f : SUCCESS in 13m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6096a11e08034ca68add2bf80c70f65d : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5bbac93b56c64e85afefa98a7acb2b55 : SUCCESS in 9m 34s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d3ac980db6e34d2f879bb3805034388e : SUCCESS in 21m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0493c10b25e4ae39a525403caf2efb3 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4c60d3d3308846afa67af8d43370476d : SUCCESS in 10m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/81b0410fc2a34b32ad9d5850e78f1878 : SUCCESS in 3m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be600e72adac40228bcceb6d43666af6 : SUCCESS in 2m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82439f94491f48c1b04f400dedd24afc : SUCCESS in 32m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a8815a25cd1a43ea9b532b521a9a862d : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/574ca6358570458da42773b1fc68821b : SUCCESS in 3m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86e6ad8ab7f14aac9ad95f9f072474b7 : SUCCESS in 3m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d1f8c107962142e98d4a278ca0b3900c : SUCCESS in 11m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7a536f8a368f455b9230947849c5bb6a : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62d4374283d74b93bc9374c8157910c7 : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/845785c008064bc6ab61a19edf912ba9 : SUCCESS in 1h 45m 40s","accounts_in_message":[],"_revision_number":2},{"id":"e8938be7c1044a4945a48b76c6098dd9e72fe918","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 09:39:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24b23128f38b02e2cd7c84fbdc97258923014a91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 09:40:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fef1306555949529ced7d6b55d7c467\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8325dcd036c49f89ee70d76039febb1 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2d54dcb4c7690409346990ff2ba735412a9798c","revisions":{"1cc022793c1226fe7273a5a01655dbd4cf3254c6":{"kind":"REWORK","_number":1,"created":"2026-02-27 15:32:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/978217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/978217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/978217/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:22:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:22:46.000000000","tz":0},"subject":"update constraint for python-vitrageclient to new release 5.4.0","message":"update constraint for python-vitrageclient to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: team: vitrage\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ieb431479ceb580455103b5e73779acc5f7c187e4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic102aa55f3460b4e03455d36c2d877243c38b9d7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cc022793c1226fe7273a5a01655dbd4cf3254c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1cc022793c1226fe7273a5a01655dbd4cf3254c6"}]},"branch":"refs/heads/master"},"d2d54dcb4c7690409346990ff2ba735412a9798c":{"kind":"REWORK","_number":2,"created":"2026-03-01 05:48:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/17/978217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/978217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/978217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/978217/2"}}},"commit":{"parents":[{"commit":"6a44666b5593628692161003895eebba55b0de02","subject":"Merge \"update constraint for python-troveclient to new release 8.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a44666b5593628692161003895eebba55b0de02"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:22:46.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-03-01 05:48:08.000000000","tz":60},"subject":"update constraint for python-vitrageclient to new release 5.4.0","message":"update constraint for python-vitrageclient to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: team: vitrage\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ieb431479ceb580455103b5e73779acc5f7c187e4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic102aa55f3460b4e03455d36c2d877243c38b9d7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2d54dcb4c7690409346990ff2ba735412a9798c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2d54dcb4c7690409346990ff2ba735412a9798c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
