)]}'
{"id":"openstack%2Frequirements~978228","triplet_id":"openstack%2Frequirements~master~Ib5a92175237d69ee094d77373b2c897158fde91a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib5a92175237d69ee094d77373b2c897158fde91a","subject":"update constraint for python-barbicanclient to new release 7.3.0","status":"MERGED","created":"2026-02-27 15:53:55.000000000","updated":"2026-02-28 00:45:52.000000000","submitted":"2026-02-28 00:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978228-new-release","meta_rev_id":"29b4f995c8c5330ba788539edb1370eca1f7f62f","_number":978228,"virtual_id_number":978228,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 00:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-27 21:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-27 21:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 17:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 21:01:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0523c5632d1a125aa2e8b471b3c6fb1a7d667d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 15:53:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"818b936b49c8c9161215ae88f9d6ae564be9c48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:02:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8dbf71918b242179e6bb5ab71407dcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471c43a37b0345c49842975b52814cb7 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd14ec9ec1f4fb28f30ad5cee0b1946 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51accdc09f74444e875e0df86aa8096b : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4feefe7361c5429cb4c451d7ee06faec : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/997e13a009554984aa018ea11317d654 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c4a34fb94584dee8cecf0dd9a1aaa89 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53bbe6ab3f45449f823c3d0358c5b78c : SUCCESS in 3m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2213a5e10ffa45b591e0de85f6bf4ca7 : FAILURE in 3m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ac5e20843fdc469aa3448b42ff603f8c : FAILURE in 5m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3e2aa320dddd4f4887b8850e877cee5d : FAILURE in 5m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ffdc81199f4449b4a55814d9f9fa1f1f : FAILURE in 2m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5ae2849edaa342109844e1fd09f6a2a3 : SUCCESS in 6m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2da2d9ce930445a483414a78e199b44c : SUCCESS in 5m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b05f166b05894adf87eacf5df3855773 : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6ef78c58f37942ddb9ce608115c31abc : SUCCESS in 16m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/242f0407575f4d8b9e467778f7e428e2 : SUCCESS in 7m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9a0d81867355486fb1d9f82ca2cd7799 : SUCCESS in 6m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7febab9a87b44c0ca5ec0b3532fe34f2 : SUCCESS in 8m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b4893c2d0f284ac0a14e49d6e3fc6c68 : SUCCESS in 17m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0230b894ef32420692a446716a83492d : SUCCESS in 6m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/aa125e35fe8d4f069e817f9f1dd6074b : SUCCESS in 11m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cf7fcee4503b4b17b85d6962a4da7bd8 : SUCCESS in 10m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3ec6d65f454741f7902ee1e1ef4668c9 : SUCCESS in 4m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3232d8c5ac004bbf8b4c759fb6df3c3d : SUCCESS in 9m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d659a284edf64cc3a0d84b55546fe435 : SUCCESS in 3m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5b346449000146e8ad4066a7890b1de5 : SUCCESS in 5m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5cb7adc9476542759d69c7bdfd5bf783 : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fb4e396a95f74d0eb1d844048691131a : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/33f2afc918f94f96b795d9436a68d204 : SUCCESS in 8m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e9f7de426e3d459ba03e2ca2a8abbb80 : SUCCESS in 17m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cffa25c342554a4aa1430828e912779f : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02a666820eb8461dbd5042462b72d38d : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e190bc033dd54cc2bc591c251affc289 : SUCCESS in 27m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3fffe50ca7584dd6a7db9c1386fcd563 : SUCCESS in 7m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/51ef8c1f79fa4cd499aa2e6b3fc20575 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c95ef63e557a466586d05ffbaac39669 : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/37585dfab8d94265b3d604c77452b608 : SUCCESS in 15m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/08d6fb74eb334d20badb30fb0a164784 : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf74d7cb5fdb4fc6aeab58fa815f54ad : SUCCESS in 1h 00m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34d9d082f598461c800ca21e85abf52f : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"08587c48ce90e313c1cfad8ee4ff29a03824d43a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-27 21:01:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"290beb37c4a41fb3b50c74abb99497d0d4d55f9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:41:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9bef26a97342f6bd32e7a12a7125c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4746946867491ab60a717e54301943 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680d468735e24c39881f19f2354eab29 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc68c97c9839435ea99bce6e7bbc64c1 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77704b73627d4c4ab879128742dbd5fb : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66c05d27417b47409927c11c0d12ed06 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e0d9cd62221473e8733f68893794996 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dbb2a8ea3ab4e82a65e7fd06e8b0d3c : SUCCESS in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e5c0d0a51bc448a497e898915c28cf08 : SUCCESS in 3m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2c6c8a6a32ba4f1ca4c0ceb14f677b47 : SUCCESS in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aa1f4f55f40c4107af801f05c78a8601 : SUCCESS in 6m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b569a131fc284aeabe446fb7529ad201 : SUCCESS in 10m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5c16fff614564ab1b5ad457dee859cc0 : SUCCESS in 4m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ec076735c474488fb82148160f30c519 : SUCCESS in 4m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/262735d91fcf4aae95006c828cc62d40 : SUCCESS in 3m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6d1cd2341053497d8dd43f4707172850 : SUCCESS in 7m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/37ff123d3ecc48aead697e4ebb5768f4 : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c8c23e701b624f73be71afaeb62374fd : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/78704e2c17c048aeacd4d80a04b67289 : SUCCESS in 5m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8fccade48c304b9597bb3cf97bcd0551 : SUCCESS in 16m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d99bb887d4714397a0e265a35446b704 : SUCCESS in 8m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c8c8a99efcbe40aea4c7d9a4874ada3f : SUCCESS in 8m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f4b2ab2040d141bf9c4fa6f3b894ba6d : SUCCESS in 24m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/376e0c9025fd4bf89b813bf73a585546 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/beeb3a4c12f04dbd851936103428452c : SUCCESS in 11m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d1ace0d2f95644cd9f2fa4693c4d4a63 : SUCCESS in 2m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/03c2a1f6c01b4fd6b84577a49e16e84b : SUCCESS in 6m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6dd81b78ab1c486bb4a5c1885ba6ab16 : SUCCESS in 11m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/81c1d78338734340b516ba14515c7271 : SUCCESS in 28m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f057b003a6c4f1cbab76f94c09aa92b : SUCCESS in 7m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0b384daaf16c43058164d9c14509ea3d : SUCCESS in 14m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4aebf34ec7c94895bcfa0320be9c7265 : SUCCESS in 3m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b227f09af32481dac5759965294fea2 : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e31f7fdd0854c6191e1257f4970cc81 : SUCCESS in 29m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fff574ea69ee4d8d86fbd42662fa1ac8 : SUCCESS in 7m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/62f0efe1c925499189699573fd985b3f : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8eecde544af84d2b8eaf09578fdb7a0f : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fe92dfd27edc481eaf2a34788d4caf63 : SUCCESS in 19m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/87d9c22bd56446ea9236bbdf46f20939 : SUCCESS in 4m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d6e72e268f4d1ca7e26f2591d38a50 : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9e3bb33f77346728d26c0e93da32d94 : SUCCESS in 1h 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6b501de285bea037e5d4652157b7513095da6d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:41:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"407b820f7f731e59dc636cd17205a12b5f46a411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:44:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203b0c881ede40308aa41a465f6c83c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8961e6ed9804021bf408cbe944a69a4 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bea2119e29449b5bcdba07e719ff161 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e681a66dbe33478099302a22287a992b : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11413bdc37f04039a8c362d37bd5f2a3 : SUCCESS in 2m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c18a52e49a484ea8a325ad98fc95d9c6 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da103f953636404584b4eb9bada22e3a : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fd8453e86d394d7e93c9bd029d378cc0 : SUCCESS in 4m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/27abebaabd5f447f9072a3bb6d04a1ec : SUCCESS in 4m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/938d4943eeea410da122202b691fd476 : SUCCESS in 8m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a64c36725a2f433983eb6a347c93d310 : SUCCESS in 2m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/83682102e9374a9f96ebaf690fe93700 : SUCCESS in 8m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f590139b8ed64d0f88c497a3408797ba : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a6a96c6a5ad6453b8591fc1d17a56561 : SUCCESS in 3m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a52ec00edd2241dfa03e88556a7f52a8 : SUCCESS in 3m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2ef14c5edad44d5c8fd380329da8b42b : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9d769e296824279a94c0d875db41cfb : SUCCESS in 6m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/889c0ffb805e45499e379bdce6a40981 : SUCCESS in 6m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dccdd3f3408442bf977ff49ea91d5cba : SUCCESS in 11m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/698ea0577c544f99b1c5f463f6ff2b03 : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/44dcd9e3abfe4543b07287acfb0ed96c : SUCCESS in 5m 58s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4913c50348e44bf485ee0e8ac013579c : SUCCESS in 18m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d3f2cd6cebfb43bf902628b30c0afe61 : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f9020f897245487980d74da70940935f : SUCCESS in 16m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2f8cf29c650c4d7aae2ccfa1efcc886c : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8b330222d984a3191d8842007a3990f : SUCCESS in 2m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/594e7d6e8e904140806dccdb53745693 : SUCCESS in 25m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/eee8e46f34214ba1bc80db9c69a2c5af : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6049102fe464f6eac188494afecb38d : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a82bbd9bf64f4da0bc70615f2593bb36 : SUCCESS in 7m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5685eaa513a74c75b08307cc6223072e : SUCCESS in 13m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f69aa4aee5a240198b486b38839c9856 : SUCCESS in 5m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93b45e0a288b4e09828cb7fb65bb0345 : SUCCESS in 1h 12m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d016d715356541a69a9eb55c312b6efc : SUCCESS in 1h 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e685f801c2aaf25c58b8eb9e5f7936fbfa88d683","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:44:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"29b4f995c8c5330ba788539edb1370eca1f7f62f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:45:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a6266d37f2435f91626829b25c7574\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb77159782b34184a3ad1ee881148ec9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ee7720dc3d66af90242c9df78773f4a63887803","revisions":{"5ee7720dc3d66af90242c9df78773f4a63887803":{"kind":"REWORK","_number":1,"created":"2026-02-27 15:53:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/978228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/978228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/978228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/978228/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:43:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:43:54.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 7.3.0","message":"update constraint for python-barbicanclient to new release 7.3.0\n\nmeta: version: 7.3.0\nmeta: team: barbican\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8b5ff3b86c02ffb1f8a036a57badcc65b58655db\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib5a92175237d69ee094d77373b2c897158fde91a\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ee7720dc3d66af90242c9df78773f4a63887803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ee7720dc3d66af90242c9df78773f4a63887803"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
