)]}'
{"id":"openstack%2Frequirements~978229","triplet_id":"openstack%2Frequirements~master~I57abfd6b1f021f47222af6e40fb5834e7e3b96a6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I57abfd6b1f021f47222af6e40fb5834e7e3b96a6","subject":"update constraint for python-watcherclient to new release 4.10.0","status":"MERGED","created":"2026-02-27 15:55:53.000000000","updated":"2026-02-28 00:03:13.000000000","submitted":"2026-02-27 23:59:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978229-new-release","meta_rev_id":"9263ad62491e392840d9dfbf55d8012cb8b45359","_number":978229,"virtual_id_number":978229,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-27 23:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-27 20:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-27 20:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 17:48:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 20:25:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"01c173148ff18e5192b718c2aca9e10e13f2d889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 15:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a7e18378bd6b671f9417efc108045105d450371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:48:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61d801ff5d82438ea10583279eb7a9e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c20c7c9d1c44a6d82b4f5321046eaeb : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24bacff340ca475d9286db30770057f8 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4369fa576eed446fba0e9430958d6350 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5646568f5a484710a7693cbfd5833be6 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c94a41935cb54635adbb89b492a66c32 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6512cd7b1a141f5a55251f2604cafdd : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30cead09eca84674a1b26a021b041f44 : SUCCESS in 4m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0ebe33bb0f3e4494b347f5238f894146 : FAILURE in 3m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fabbd1ceac244de69360b7655cf80e63 : FAILURE in 5m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f6f5e2c51ddd4cf1b62c691e7de00741 : FAILURE in 4m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3e8c85c3cb0d4c64890674c140845698 : FAILURE in 6m 22s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b568390ad3d74770a114841c3a1fe159 : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/18ea6e5ee2be4af39537619f8f730437 : SUCCESS in 7m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fcd0431bdc584e89b8b8977c45ce32bf : SUCCESS in 6m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ec50234c794246718e9bb9252c795305 : SUCCESS in 11m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/51f0f767e7184f8897dbfb1527c1733e : SUCCESS in 5m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/184e22ab8b8442f6b9395ce842c8aa91 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/428b77baffd247049bdc4039e1219416 : SUCCESS in 7m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/036065950f654f4f843767fdcf850a7b : SUCCESS in 18m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0fc6dae1ffc44678a6e71d8af743a23 : SUCCESS in 7m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ac5e7b587beb42348745b6e5d4a35cde : SUCCESS in 10m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2a7bc0522b2f440bb8d69b9fd8bd11d3 : SUCCESS in 15m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7f7074d0c97a42939c539e06f4922761 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3d47b45f84ac40d0970e3385eb9e203b : SUCCESS in 12m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fb16cf6c007c4e50a11e28be11a45161 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/65843cdcac814021b244220bd2c040e9 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7e307142beca496f8301d3a817ebd231 : SUCCESS in 12m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/02ec7e1d9a8940c4aacf1af138e682af : SUCCESS in 26m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e23cd6a3d08141b5b5606ea9ea4d735b : SUCCESS in 8m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c0d63a2dd3f54fd28c32a99305fdaa81 : SUCCESS in 11m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c18d073c30e34419ab0372602abd591c : SUCCESS in 3m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99695911f0ac4e4bb401f0298b511f1c : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce06693de95c401798daf0f42fd55f91 : SUCCESS in 32m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9065d2eeeda340a789e752e95f4f7c9e : SUCCESS in 5m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/520baf4fbfb84db5855b68759c6726be : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3765bf3c317344cd91ae710a3152ce68 : SUCCESS in 6m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/39a7a2c9be0c4a61832abcd30178cbdc : SUCCESS in 20m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1479f3e27d9540a6aa2acf60580d0edb : SUCCESS in 4m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4639cc1a77e24676908a37af26e24fa8 : SUCCESS in 40m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc62bee02baa486481d0f4429fa31c7a : SUCCESS in 1h 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8efb692e36e7bdd6ae9634e8d23038b3c3ebed56","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-27 20:25:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d8e810e8fc8f0a3e37529411b6bd85cb2e1c38e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:07:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ee90d82fc44c5696a1926e32439c06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ee17959e0145a98a9278d68741e877 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613e9b5811b3450bb35e2f8f0983381f : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ade7c63464a49f495a9c018eff14ec1 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ea4d09c17e248bca2381772dfb00e28 : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44460bdaf84142f593f9aad4c4059add : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f27dbcad327043c5a509c3758ecf5715 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6a83af87b5b404e9c2e24ba40c5c7bf : SUCCESS in 2m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f06bbc689837474da18b76cfa6f86ef6 : SUCCESS in 5m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92722b57211545e5913b6bcbe1f8c012 : SUCCESS in 5m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b58a749e31154a8aaa155bfda82cd04e : SUCCESS in 5m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e91cddc8ec21490bb0faa4acf099ac2d : SUCCESS in 5m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e9405645a7ab448c9327e4c74e886885 : SUCCESS in 3m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c02d763e9f1a4e48a5f7eb5b47a8f391 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ca25994d2df9424fb2078c554c00d611 : SUCCESS in 6m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bd93e3e0b3b04577b1a5d8d6bec12120 : SUCCESS in 13m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7757d8658d444c76a5828fcdd3f52106 : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/45016b84acb74735b5d61a06a6bdd28d : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e2ddec7ea6864d73892eaaeb240c2c33 : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f76b6b4b4fc84d06916ebb2270503754 : SUCCESS in 17m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f40d2368e244f3dabb99f3c2de9d878 : SUCCESS in 6m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0e18f8e3e9b44ce8ba7f894454cf32ad : SUCCESS in 9m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/035c102b9bcc43dda0b80e0861ef5f12 : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f89b8728810e41df8a36c5b7af533742 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aa26ec7b6c834f2abd4e9ec5941f3ddf : SUCCESS in 13m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/105ca44005634523bb8480e2d989da91 : SUCCESS in 6m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b9a1b2fa6db046d7870c9afec248c39a : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/60161333c7d74849a8ed76a72c994cd9 : SUCCESS in 12m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cecf65503a9f436d9c82fc67dab3b768 : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5cb74d27dd804311bf72fd9e03662eac : SUCCESS in 9m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f807c94223d84f61807eaf916f5e7b07 : SUCCESS in 15m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b39bfb5f69f6410b9d70d09c697be0a8 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7ba44fafc99446686aa6f084487c73c : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32850a74410f431d8dfcb844a418cfbc : SUCCESS in 24m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a307aa30120046b2bd8c3b76dc934082 : SUCCESS in 7m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/76d2051f4411423d997a509454a5f2eb : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94f2717c30ba4a34a199f105872b61d5 : SUCCESS in 7m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9f73d28d4b6342d6a660f294bb420ccd : SUCCESS in 15m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cd14bc7543544930a2731afd337d7c7c : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/937b22f8b4604d5fb1360fe18226b89e : SUCCESS in 45m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/745d2bf718b0423ca383a2916db5b119 : SUCCESS in 1h 24m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f7c15d3582c08b49dbc96b4f60e6f0dc67ac1372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:08:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5ff7c76a9b9ab81d26493922944a1e6f8112f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:59:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c411a19d2044b883e9950a59bca327\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e7f39b056248979991d6702cfc36d3 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427f9096cd5e498993aa18dec226c18c : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef37f0bf34c34a6a9322e57721970f5c : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6414daa826e48b1bda8383bb3f3dd00 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76964b2ea98440258ee06a7bd97ceb92 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c68c4a8320d476390e55566f4d3ab85 : SUCCESS in 4m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/15b2968a490e48c49eed00ad0c61fce7 : SUCCESS in 5m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/928bff9a28c646cfa1a380094ff33d79 : SUCCESS in 6m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7765ee6a04c44ca7a2a9d1c0dc0bb77c : SUCCESS in 7m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bc86f6d90f374d66b8c703feb7922cd7 : SUCCESS in 5m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2be78da82f414876bd2445ba8ef56422 : SUCCESS in 9m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/edba9e944f6b42a3bfb92266ac99e87d : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/df9112603cdf417993502a466804ac86 : SUCCESS in 7m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/186f5a99e131497d8964aadec609e640 : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9eab424a69104fa8a6966e51ce7b4ca6 : SUCCESS in 17m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/806f7381316b4b95bdae3ae055b0c350 : SUCCESS in 4m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/20eb364a00e940b5b4b00882116840b7 : SUCCESS in 7m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/eb2f42fccc624e06b8025b9e07fa606c : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/206d91e636654f3e8bdd6f94aa81b00a : SUCCESS in 3m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/82b3443a6d16483c8e0748a65d475bdb : SUCCESS in 12m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1623ae46a76f412ea71f39f4e55058d8 : SUCCESS in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0a5c14ce4c704000a2b88e5415069564 : SUCCESS in 8m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b0c68677e3e941aea727c108655c60aa : SUCCESS in 12m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d5981d53ead04b1a9273ded336252b44 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2dad033bde9944d7908feb7a6dd54573 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25929a1a252448de86152b20f13f5c3e : SUCCESS in 29m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/481c862986a4477981001cc1e792e110 : SUCCESS in 5m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/681ce3b2e4e74c5cb2e35bfdc39fbaf0 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdce0905a40f4ef8bb7e53907863d3d6 : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3b779260892841e29c9f35f134457116 : SUCCESS in 12m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7d25c621ea9f47bd82150e1df4113778 : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ab8ddc6fc9b400cb4a77de14e1ffc13 : SUCCESS in 1h 09m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b555a08521d48e1984ed7be346e6dde : SUCCESS in 1h 44m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7d9cb8c1d618e8cec00d3dab3260255b9c43b163","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:59:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9263ad62491e392840d9dfbf55d8012cb8b45359","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ac2f3b20da4e70b3d4639f08d60fea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49189c441574488f9981fa837fecb13d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a9bfa424571226280e73af4121193248404c68f","revisions":{"4a9bfa424571226280e73af4121193248404c68f":{"kind":"REWORK","_number":1,"created":"2026-02-27 15:55:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/978229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/978229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/978229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/978229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/978229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/978229/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:45:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:45:51.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 4.10.0","message":"update constraint for python-watcherclient to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: team: watcher\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5f2778f889103cb648ff1a2d7f6f3f5647e8988b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I57abfd6b1f021f47222af6e40fb5834e7e3b96a6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a9bfa424571226280e73af4121193248404c68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a9bfa424571226280e73af4121193248404c68f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
