)]}'
{"id":"openstack%2Frequirements~978230","triplet_id":"openstack%2Frequirements~master~I6354aae30842847572edf2aeb74beaaf25429a3d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6354aae30842847572edf2aeb74beaaf25429a3d","subject":"update constraint for python-masakariclient to new release 8.8.0","status":"MERGED","created":"2026-02-27 15:57:20.000000000","updated":"2026-03-01 07:30:10.000000000","submitted":"2026-03-01 07:29:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978230-new-release","meta_rev_id":"a46b8fcd0776156977d8ca6a93de76debec28e2a","_number":978230,"virtual_id_number":978230,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-01 07:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-03-01 05:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-03-01 05:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 17:17:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-28 13:04:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-03-01 05:44:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c3dd72f500fedfb022b3b4b2ffc72aef58bfc01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 15:57:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63cfabca21c292f50bbb7ffe90f8e39f9546aa8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:17:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14d6ad52b5b04b88a322369b290b0193\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1730112666d248ea9de1ab21a5799129 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc3e1ccd40d42cbbcf5e2b1f1926468 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/114d11a55eda43d78af098b10c910c3b : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1199a39df5904ca3a93091dc229901c3 : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/245a2f7329b44b8593b71c0eaa22b264 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8f768db04e04c3191ba1b707030efa6 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c47a2afcb9fa4c74895d60612e283937 : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1cfb8dc74ea441438e72ea3c80503f1c : FAILURE in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e8503f1ac7a24e08bba35a070878b007 : FAILURE in 4m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4caaf106bce44a1ba8759f023d8c09fc : FAILURE in 5m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/63b4b716dfe04851a1d99b5fd985893e : FAILURE in 3m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9ccf30ce81c54c54b44437ae5befc763 : SUCCESS in 7m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3101828c67b94c39854d9a9a7b0cda64 : SUCCESS in 5m 37s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/651e84e8c55540da8cb5c2527219f961 : SUCCESS in 5m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/46227de0aa0f42d6a0cd7ff786958f9d : SUCCESS in 11m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c286c202bb2847b3a0b042a8f4d3ccd5 : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5fb22c0713b84388932f4b5a00ea5d6f : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d7699335ffef43cb874d4da08f6cab5c : SUCCESS in 7m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/17bc2cabefc7461e8d59734cec021d1e : SUCCESS in 18m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67a440bdb62443d294bbc4bac2098e8d : SUCCESS in 8m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/03710dddcc5f43709b94c4e94a1a37c0 : SUCCESS in 8m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8952c2d11a3d46a0af13355643e31a71 : SUCCESS in 15m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f822ad4f37444e52b016f1c4fb2df3ed : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7a9b85a4c5c644f0835cbf213868c025 : SUCCESS in 11m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6f2f3dd7d5ee4e1eb371bca06ea63864 : SUCCESS in 5m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/aca84376fa1e4603b2679abb97f21c90 : SUCCESS in 7m 05s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9bbe0e933927470a9af05163b48bb328 : SUCCESS in 12m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e74967951d9e491886bd7c7d603b0cf5 : SUCCESS in 24m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bca50fb0668e49adadaa74284a6bb88f : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fee4f4ab4bea4526bf7347932ed696ff : SUCCESS in 14m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/53ee242844ea4d4c94eb80dbd7ea46c3 : SUCCESS in 8m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96335f83010f4394b0138be6713e3fab : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b20cae7af314674a87a455e059ab5a6 : SUCCESS in 30m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8dba8cead2f14415a8bb3a9d4c023d69 : SUCCESS in 6m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9f8a7d042ee74f81aa0d9aee5d6da1b1 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb9ec338db1746bb88f05759c4565a67 : SUCCESS in 7m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/87822ffe395e4772be924ce470762ee2 : SUCCESS in 10m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4b74bfc7c4ad41c08200dcdc40e78bc1 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d0f89b254834d62bd725acc1718afc0 : SUCCESS in 1h 04m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ce0944c0f104de39d8fa70ede97f112 : FAILURE in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"49388ac8a6d8a991a449749124a2d233f65becd0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-28 13:04:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a26abb607b3ef45c5a86c0705c1e9cb5b8f6115f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 18:45:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93168ec03094f669693af8186a050af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd9da093928416c921ce09c2a15927d : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811933a8d8c146f699cc98be1165a2fe : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5e87a51193d40b4ab6f22ca61fb9704 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/533986b12e1b495c855a7109cd8a066e : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/840144e85ac943bbb0d2119dfefafe96 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e5225132dac41a7aa91f2134432694d : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63b3134934fc4cc38cbb3a39f9326c8d : SUCCESS in 3m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9339077e676743e69a26cc9b025d8c3e : SUCCESS in 4m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/39dad5e15e8e4443ab9b7289b6bc4a41 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/247bb286b1cb42b09d2c2904e6a851c6 : SUCCESS in 6m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/20467f9f8bd1497480d798544eb75a79 : SUCCESS in 5m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f9c62262d9ef47fbb73d2c52fc33669a : SUCCESS in 4m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d1006cf8e26244119906d43f9abb267d : SUCCESS in 5m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/026e5e6efe354b24ae9e0b112d9e934f : SUCCESS in 4m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b77339583cbe4f8ba9a6ba7cf312a2eb : SUCCESS in 11m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/89e91927e4c14f7ba8dc51b3820fe828 : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/99b1c662f9a94ee49ec0871fa58cc66b : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e3afea4743c842928e6d9a2a684b22dc : SUCCESS in 7m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/35d5f8cef76d45da9109f1317a4d77b8 : SUCCESS in 17m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64d37075b8f44aa2a1b4dfd9e437b3b4 : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa0206ff9f9b4ad0bb85b95c6acc1846 : SUCCESS in 8m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ea1d2a6c83c747da8e604f63488aa374 : SUCCESS in 12m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6ef76f9ff4f14991b7ed12e021c11517 : SUCCESS in 2m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/161496cd58b24b72a7eb3c4aa88f05f9 : SUCCESS in 8m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3e52f4ff38b24e59bee943b5f4aa7962 : SUCCESS in 4m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/904f39b462384a5ea05e051531b3b216 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cf24ef0ac44b4c8d92b3ad41d4f5f89f : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e8c61a8894a54125b67aaf38586c8ae6 : SUCCESS in 27m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ce32514c5244465bb5285f50f2afb9f4 : SUCCESS in 8m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e642936a9fac402cbbeb69ed093c0e5b : SUCCESS in 17m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9b541d29cd3d49d29705f1274fa34418 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/707c9dfcfb2a49b4bb7c01acdcaac03a : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c69b42a8b854444a15192d606728a35 : SUCCESS in 27m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9de89df95e2c4fde997285a6d7b07ac3 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fd31c87b6bd9449aabb1d7ea8b1da14f : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35d85ecb4d8f49628077a1b1af6e7247 : SUCCESS in 6m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6cdf1b9b41ee41cd82bfe6e2e00aa35c : SUCCESS in 20m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6199fb64aed34c4798c7a5dbe4b375f0 : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/876dd1fe4063463e9b2cfbbaaefdcf05 : SUCCESS in 1h 05m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26d5611d737a45f2bb34c13b59ea068a : SUCCESS in 1h 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2f4ad0910ab8807a645b8b60323f2dad337fe49f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-01 05:44:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"790a8f0183fed915559fcd5881ec57be12aee478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 05:45:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41049e651fd6609651123ea3a464371e761f8ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:29:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a5747296b84ce3a06ab0dd1186d407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7cd7377ea7b4f09ab7d624e9d045d6f : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e105b3be1c4598ad2edfe75947a035 : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/537175b45c564d6286772f6eb2d6d5a3 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80d28a4a97be414fb382a1d40e2f9f0b : SUCCESS in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c8e3a41c8584b0b9b6e802d47bfcb74 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f972d3c6addf4a04920067afc97d7d05 : SUCCESS in 4m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e9439eb8ec80438fb75b2e34507f042f : SUCCESS in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/900d498b966d4be2ac72af73448fecb8 : SUCCESS in 5m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/33ab4289e84a495baccabcced62c9294 : SUCCESS in 9m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7efd828152b541a78ed38d9eed84077a : SUCCESS in 5m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a4398c8af34843fd92faadcd24bcd8ce : SUCCESS in 10m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8df5266e5449478f9397f599b6c116e0 : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f42fd5520c3547a3b8117b09012f4cad : SUCCESS in 6m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8cbcd796d02143fc87b3419e510654a5 : SUCCESS in 7m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a21ed82ec87f4e849bcde598deb6ad91 : SUCCESS in 7m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38838a138fc74c549bbd129e26af102e : SUCCESS in 9m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f0bbd60bfda14e639327eeddc4c14d9e : SUCCESS in 4m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/81b29719f3b44ffbb9a77196632439ed : SUCCESS in 11m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b7fe9937626046eda1056108798e61fe : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2ec0fc3679a34b329812e013ed641a19 : SUCCESS in 11m 54s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e0f63f50548c4cc7a0041969a9e01a7c : SUCCESS in 17m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/973a3f524e9947fbab8c8acf96931755 : SUCCESS in 3m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/659a529211774450ab550f69cbd155cc : SUCCESS in 18m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/65d703f4e51240e88bdb64fa9695f4c0 : SUCCESS in 3m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3bcc2308a50454a8c0c97f1d4c2edc4 : SUCCESS in 1m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7614cb09997f4f1ab4f2783da0f940e9 : SUCCESS in 21m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dd6debf4dc514da18c53fc5842f16c72 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0965a58d420c47428204452d7b10fa19 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81cc910eb11c4223825480a8bc65200b : SUCCESS in 6m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7cc438292a9d4cfcbf1833c2156105b5 : SUCCESS in 12m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/933e16442d86405cbed59b376571debf : SUCCESS in 4m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b39af800ee94ca6b705e8289fec3de5 : SUCCESS in 1h 06m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99b0e903fd80447ea038eefb513877c3 : SUCCESS in 1h 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"762bb7281b4d9292f9f38ac16c6f37bd1da47e87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:29:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a46b8fcd0776156977d8ca6a93de76debec28e2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 07:30:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c61ce6973575404084b2759ab2d6831a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c1e2aee00d474e95ef52202c8dbbf5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"218f22ac3590cb7a09c31b1b25bf562a560ed32b","revisions":{"218f22ac3590cb7a09c31b1b25bf562a560ed32b":{"kind":"REWORK","_number":1,"created":"2026-02-27 15:57:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/978230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/978230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/978230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/978230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/978230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/978230/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 15:47:17.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 8.8.0","message":"update constraint for python-masakariclient to new release 8.8.0\n\nmeta: version: 8.8.0\nmeta: team: masakari\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia63370a0d3389d5ed4cadae0fd647bbb2889d870\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6354aae30842847572edf2aeb74beaaf25429a3d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/218f22ac3590cb7a09c31b1b25bf562a560ed32b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/218f22ac3590cb7a09c31b1b25bf562a560ed32b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
