)]}'
{"id":"openstack%2Frequirements~978233","triplet_id":"openstack%2Frequirements~master~I8b4ac7380a01a414d604409eb976cd951b977507","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8b4ac7380a01a414d604409eb976cd951b977507","subject":"update constraint for python-magnumclient to new release 4.10.0","status":"MERGED","created":"2026-02-27 16:13:55.000000000","updated":"2026-02-28 00:45:07.000000000","submitted":"2026-02-28 00:44:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978233-new-release","meta_rev_id":"612255b5b27a6691b42d697538c868f6133d8bf9","_number":978233,"virtual_id_number":978233,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 00:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-27 20:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-27 20:26:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 17:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 20:26:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5b8b2fa134c55f88ebd49644d5438ba97536439d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-27 16:13:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ee080b267a71b35be74f9d36802a6618df0fcb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:30:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5168c9d491944d21bde45a3db50a517a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ecc8c220e1f4d12a2b76cb1a906d492 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e3addc607441cb985b201415111c2b : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b74713075ce642bc905015c51c3f263b : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff0dab5f577045b1bce3858b34390fe7 : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae785f9f88ac49f58ceb561d34b82480 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8162a58193944f56947d164a864c02f3 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/925c252e47364600a51e91d2a7f43df2 : SUCCESS in 3m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/00a19052ab8a403dbf8da17daedbcd3f : FAILURE in 5m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/89f42ca6b7aa421ba2292d2e45fd556d : FAILURE in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/697fbd3082f94b43ba21884647308509 : FAILURE in 4m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/323a0bcb3fa447e7b3dd60f51b1442b5 : FAILURE in 3m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c207d8e68d8849c489dd1c1a09dd1c1f : FAILURE in 5m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bcca8c2bc7e94fb3bcfebc2b6e027ee9 : SUCCESS in 4m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/512721d8719e4418a03c2fb970a3f8ff : SUCCESS in 5m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ae511ee57f3447589b69a6274b95bb8b : SUCCESS in 12m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/709fb65978c849aa80fb76462e893db6 : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d1a5bd50af874f9dbc506b21e8b4d666 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e59c56c6eac0435a9619152eaed01556 : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6f240c0755c34a99a7eb266a0ba42c68 : SUCCESS in 17m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/363a1916f74444fd8a985db483518b82 : SUCCESS in 6m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a5b8a6ec31b44b669f17f2d3acc50d10 : SUCCESS in 5m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2ca25bbbc310481dba4a66014c4f6785 : SUCCESS in 12m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/04f3ae6678604f7ea4d7e758152db70d : SUCCESS in 2m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/542a4ef764d64ff8a08a187d5c6dcbf0 : SUCCESS in 12m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/55e6e57aa3f94e94b0790abf3040c1a9 : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/481e71234b5545b9af5b1f748f2b7bf2 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9904d4d676114b3ca440d5c7db44b7a8 : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c5e0dd6dbc0c44ef97edda5b10d1798c : SUCCESS in 28m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efc6357aacf544dca942a257e28f07b8 : SUCCESS in 8m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/361bb48c8bc34e5d8e590d7918ed8542 : SUCCESS in 14m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f34b0a0e46494156890519d8a5e1c25f : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dab35aabfbc643848b19d30d2c82ffb8 : SUCCESS in 2m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83a2919ec23943f9b861c8e6b23bfc5b : SUCCESS in 25m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/03708258f4de420ea068b1d44451c756 : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/02331282f2d24f9c89e78e64a7bd7b43 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58862525e7c448ee81091c75a4942435 : SUCCESS in 9m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/53f7e80d8a6b46d79b83a483ec77ccd0 : SUCCESS in 18m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/708c9d7047bb474ba5df671712383a6b : SUCCESS in 4m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a9ca603c9b4471ea7aebfa9b967c156 : SUCCESS in 1h 10m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e38d1c6063574cb9ab6dae8696c6cb3b : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3bf782b669d57bba4b42930dcef6522fa26c784d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-27 20:26:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b5f78f153545b86d38655124df0164d2e39ac3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:31:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b73b1524b54d2a9f68b855ac86f556\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34b9de50a0d4ac0b772ded7646457b6 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c0aff654cf4b12aa193d325ac1278e : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e55703d56b9b4a4ea93cb24eae2b3d64 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e7f6abc863147dbafdd28326f614bad : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a03bb6c27e2402eb995e37058bd959a : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2934baade12419d8a4a929b4909729c : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10d758088a394868be6c6c209a85146b : SUCCESS in 3m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3e75580f556f406cbe6e5a0cc4612d3e : SUCCESS in 4m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d92d73337984465b900208ef72de80af : SUCCESS in 5m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a1b3376bbc884607b1b0725622dc5d57 : SUCCESS in 10m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/29837b038992488891650fba494c7cb8 : SUCCESS in 6m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ac37f884c39c45f6a3c98578c2a8a62f : SUCCESS in 3m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/808af9fff7a94ecb9a6ba5fd91a77dd3 : SUCCESS in 6m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fb84d091287c42be9c955f3ccbdc4b82 : SUCCESS in 3m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/38fd40b6962a4b02b92255cec2268d0b : SUCCESS in 12m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7e63382381894c238c5e44769cae57d9 : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/929a0f6e32364795a594f39f181579b4 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/64706448620445f68212f74642371364 : SUCCESS in 9m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/11c3166b8d15488fac7ed4373a509794 : SUCCESS in 17m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8586d9b43bb54d638f18f08d05f5bbd9 : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/549232d3a9a54c288593f24cbc7047a9 : SUCCESS in 5m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8cc4c83b394a43638e72933005e3bc58 : SUCCESS in 13m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/197d40990a6b4784b1881d3e4b5bfd09 : SUCCESS in 3m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2e0ffb92f8284c4491303fe69158716d : SUCCESS in 6m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e84638d97f5c4373bde0221c6e83b235 : SUCCESS in 3m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6b9d8271e0814385b21b08d58618aed4 : SUCCESS in 5m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/15b18d2774dd4452b1631ce695c0e16b : SUCCESS in 12m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4d78c74bef354395991559bddf0a14c9 : SUCCESS in 19m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a08be4a66edc46d883ae96c4702838c1 : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/491de02a86cf4015b9d607f4ac276ef7 : SUCCESS in 10m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/517c7162f3bf4df49085cc55db46031c : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77af2f00959d470c95a61604b8b12bf6 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7094092a4004eb8bce186ee41b42ec7 : SUCCESS in 30m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f12ca15526854ce495cef0bf89d246f3 : SUCCESS in 7m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c1382e974c4c4a9286aab036bc606f6f : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/231729c5b39344169067a35d82d0944d : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9b01a1180fd84c529ceed766d62899d6 : SUCCESS in 19m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5cc942edafef420cbb76e580f5292c13 : SUCCESS in 7m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59a87ee21a304e7e90510a970bbe5523 : SUCCESS in 1h 09m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1043e6f1405c47b8aae585a9c337cd49 : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f2a0de1b4011ffead0f237d9f657283e370927d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:32:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8af0dec62a6b02a012bcdeec22ba4dd9d8fca32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:44:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43098f1542404c899670c7c3ac77d577\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c24ba3cba9d4536b15b0f9d0688ff26 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40c0749f9654bff98a464906ff0c1c8 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82a8770af5a1434ca47c4722d0f733c6 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88ed24a5e927405b8b392b47a35b7711 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/526f9616f1a5414a9ca4d2b6c6050313 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7122ea2b6b64b168b5317b3ca3aad5c : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6f748be6f0f84c268bc48e00c12a111f : SUCCESS in 4m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67b55447ee9444ddb26130edb86b6c26 : SUCCESS in 5m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/68f3081211644e42bbb0fb243a47bdf1 : SUCCESS in 4m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/201628fe70484b9595e57deb4d385f5d : SUCCESS in 9m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/005713481995471b8a9ff6934aacabc4 : SUCCESS in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/258ca1395a4d422baee2d53c09313eeb : SUCCESS in 4m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/009ecbcf20f043efbcaa049e8df4c43c : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ae3499decaf84c6a915edeee18126e29 : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/628e7d8fc27e452b8012c654510f03eb : SUCCESS in 17m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6f7531a1a1642d9a6b458a91bd8be69 : SUCCESS in 7m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/be9e55e39b2c46968114c3d14c3a6462 : SUCCESS in 8m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b77082da3f6a430a87f752ede1fd8b12 : SUCCESS in 12m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6db7ec4106824fb9942ad6c44f9b6f36 : SUCCESS in 3m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f9c9823d3a02491bbe5f8d5efdd13ad2 : SUCCESS in 9m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0d5d1b2524ba4455ab647c602080ea9c : SUCCESS in 18m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98b7bb4a3a4a4f3eafad7f9433c763a1 : SUCCESS in 7m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f83a40eed899484e96e573e9b19fb262 : SUCCESS in 10m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bd41b2f746ff48f9b65b2b11272d9b58 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e824153019c4ce5913491a70b922119 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/561d85bac300492386082de8f623d71d : SUCCESS in 26m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cb3574401c164c93b9c71546cf7dfe0c : SUCCESS in 7m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b2f237e3eb704bc6a5fac6bc9417fda6 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9eb45f8f2d6423686dd838cfd3e7ccb : SUCCESS in 5m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2c62445f5569458993a8b090c92b9c46 : SUCCESS in 19m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ded400f73f1b43cf93c0eddcd28f406c : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/329e572e9cd543f09dc6cfab97944280 : SUCCESS in 1h 16m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1068243c2e5d4a86b1edf72255001420 : SUCCESS in 2h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fd79a551f3a84ad27c473d0bd10f184888be1890","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:44:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"612255b5b27a6691b42d697538c868f6133d8bf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:45:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3107dd964b0d4a919a8ef5876174ad6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eeaaab147f94b04b18b0b4d9135b6f5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01bc12ecbfc48b5e94e442f5c20a64abcf096a7e","revisions":{"01bc12ecbfc48b5e94e442f5c20a64abcf096a7e":{"kind":"REWORK","_number":1,"created":"2026-02-27 16:13:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/978233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/978233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/978233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/978233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/978233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/978233/1"}}},"commit":{"parents":[{"commit":"9decec49f310d9a4344242f27191ca8954bead22","subject":"Merge \"update constraint for python-heatclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9decec49f310d9a4344242f27191ca8954bead22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 16:03:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-27 16:03:53.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.10.0","message":"update constraint for python-magnumclient to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: team: magnum\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iae2bb4af9ea6b8cfed157c3380638df44c97f130\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8b4ac7380a01a414d604409eb976cd951b977507\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01bc12ecbfc48b5e94e442f5c20a64abcf096a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01bc12ecbfc48b5e94e442f5c20a64abcf096a7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
